)]}'
{"id":"openstack%2Fkolla~737934","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I014c9a52b2f61a86e3a8adafdf4783c3f58148ef","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1885104","hashtags":[],"change_id":"I014c9a52b2f61a86e3a8adafdf4783c3f58148ef","subject":"Fix Monasca Agent Statsd service under Py3","status":"MERGED","created":"2020-06-25 09:23:05.000000000","updated":"2020-06-30 10:36:06.000000000","submitted":"2020-06-30 10:34:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737934-1593513259934-886363f1","meta_rev_id":"718507a2430eae47ab8430e19c1dc3c8217314c8","_number":737934,"virtual_id_number":737934,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-30 07:37:12.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 10:34:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-30 09:14:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 07:50:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-06-30 09:14:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 07:50:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-30 09:14:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-30 10:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06d7211511941460755f854fc773ea0fbfc7891c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-25 09:23:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1304f246bb20ed4dea9b6c7c7cde4281d48f875a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:57:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b73af00713434d8b8ca256b70eb3d9d4 : SUCCESS in 35m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cc9f27a2270a4af6ab3ecbd4acd844d7 : FAILURE in 23m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7f203c57956847f2ab3ba9aa623dc4fd : SUCCESS in 54m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0e8ed52439e4f9c8624982ff7954609 : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/29c9d309ed6043f78672c2e9455f372d : SUCCESS in 1h 18m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/79167c438774490b91054de8056683d2 : SUCCESS in 41m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2adb3b11505f4553b45c14c835678eaf : FAILURE in 33m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3a6ecef7052e477b9f9b36fb65ad50c5 : SUCCESS in 33m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/83ae3694112e462bac751925c021c8cd : SUCCESS in 1h 29m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00b57458253341aa88c0424de0140c59 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d87eb2a68404789be8e03d279b35e9f : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857d7e7ade174915a8ac5c8313254464 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bebaecc11cf4f05905bf6e05f42226f : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b14aea4d76548458ef3fd8e20cf1e6f : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84e9655c22d34ca0827380dbbbfc1d00 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/696104f2dd8f4f96bf53e270517f8c9a : SUCCESS in 30m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/563dfe62e6bb40bb8691531c99911fb9 : FAILURE in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26932b9cc57a42c8b4b95c491e1b1b87 : SUCCESS in 59m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3ba59b193d9491daffdb96ef2574e32 : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4f2a2bca508b410fa1766fdda63fe4ee : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2e0480b2468b57bc1e1b314ee07cc4453bf9e3f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:40:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb6f620e0781463c86213238b51b57fc : SUCCESS in 1h 08m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4c4695f0baa40cb8919a2188cf6b1c7 : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4396c33480394b35b43428c1a9b521a6 : SUCCESS in 2h 11m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/adeaacd9192c4960b487c6fcdd272353 : SUCCESS in 1h 08m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04cdcc55004930c76e785aed12575d0bd4b4c987","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:50:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"17f2f64b6a44fe2e58d39da712a1c5f561df2ca1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:51:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5548c8420cb0322ce642f964e48a2a5489c4eebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 09:20:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/24a672aabc2041a5ba8baae0e09ef571 : SUCCESS in 33m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a511973abbf04cc9b831f63ac8678611 : FAILURE in 26m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2070d4d330549b7bb2aa92228a2ef76 : SUCCESS in 1h 16m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/60704366db6a41ef9d83d42565b79e77 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/53b9fe334c484322ae9b8bb438b7e551 : SUCCESS in 1h 18m 54s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3489bfdb4ce843d8b229e3b32c6e7c21 : SUCCESS in 44m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c7fbce3b078340c98f88e8a5c1209d90 : FAILURE in 21m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b9c6328697a543fd95629cc1c552d7da : SUCCESS in 19m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0f50c1da01f4840b8feaecb82c9125b : SUCCESS in 1h 08m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dd15374a3754f27af7bfa3c4d250a10 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8cf833b44e43599f3eaf195de520bb : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1f8faec9b54151b1f4fce258f0a214 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7d0aff6e35e47af967d7518c005d059 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52effebaf9fb473a9feaba83b69653cf : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/319f4a16e211422f93a74272fc29b09e : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/62089b307209400f93b10cdf22ef97d6 : SUCCESS in 23m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f283c4c228844abb840d508d7cda99a : FAILURE in 25m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d6c843ff7a149fb9647b5e6a2f93bfb : SUCCESS in 1h 05m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87341d1626fe4e819d9a0506d815f8cb : SUCCESS in 1h 01m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d326a5a2edc04ed99651649d79cb9b9e : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a41d9cd02f5c25eb802b6e70c77f38c8eaf73eec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 09:41:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d6ebebad21f427fa64cc3454a5222f1 : SUCCESS in 59m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1322d9708220469294e75a540ed77d15 : SUCCESS in 56m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1dff796cb51c46939453e019e2e7af50 : SUCCESS in 1h 45m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b5ed76e07c44afe815113611f2e5de2 : SUCCESS in 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dafca8b1b0a4b6d9856226e35814903282c4a204","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-28 09:49:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70a5147bd79a28efa85eaea2226686b8ae9d3b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 11:09:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e35c260658ea4e9b90b423449ff41f78 : SUCCESS in 43m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f0d43a8d1ffc442f937348a92c096d50 : FAILURE in 46m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bdd09fba2094497f8c0a446c391bab5c : SUCCESS in 1h 10m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eff4b367c5b4457589e8bb0e5ca97395 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ae6b7ddcf231452c8c034c2727b42b25 : SUCCESS in 1h 17m 45s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4a2b48e832b94fb996eba2cf91125bd8 : RETRY_LIMIT in 6m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/98378c720066453685ce9ee4214abb7d : FAILURE in 23m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b58b0d060fce4e04b02ca22f172d6739 : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7fcb7b411db428cb2e114a9e7a9f853 : SUCCESS in 1h 09m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bae731995f3f4f21bccbadc10e112fcd : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35748af674464608961b45f130779591 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5846f74e494bb49925bdfe18393d95 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99de20839b4e4b49b60426ddf1ce0277 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262d5a4ba1664e20a7bdbc0b06851234 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c21c68f408047cfb6fc29fe225256d6 : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42907d4c1c754031ac0e06f6713988ae : SUCCESS in 26m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6b77d36d05e45dbb3a975c5cdb8869e : FAILURE in 39m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7deec71c086f496193e20540901aeda9 : SUCCESS in 1h 10m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5227db98f7348fbb95e70d0c9c461c5 : SUCCESS in 54m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73e5ef21e3224073b2a7620ef61f80d1 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8f7806175e22dd840a9e465339f1a6b54e4d9d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 11:21:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c196d58336c54282b477c424f734534d : SUCCESS in 50m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff26d46f3ae6416c998d826b90a53b5b : SUCCESS in 32m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58d14b6e49f14126b2ad5ae0b432231a : SUCCESS in 1h 26m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac358e604b6d404b8393ce01a26dc27f : SUCCESS in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88d31624ee98d70903d8062166934f53a112e7a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-28 13:55:41.000000000","message":"Patch Set 1:\n\nit does not seem to exist so far...","accounts_in_message":[],"_revision_number":1},{"id":"e3b9f2c799e14bd0923e4209cf27d63f5414c344","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-29 08:29:42.000000000","message":"Patch Set 1:\n\nStill waiting for the fix this pulls in to get merged...","accounts_in_message":[],"_revision_number":1},{"id":"3317e6c844d48544fbf35a14a2a8fbf9284704d0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-30 07:37:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de81c7f83c38413cced28afb4ef141ff8a7e244c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 09:06:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a6f7e9dfdc6948499c159bb80177abdc : SUCCESS in 22m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad5d7a528f874fdcb04dd7b049dd6b89 : SUCCESS in 23m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7ec501bc6c314a3b9fb1cb9734b3ef8c : SUCCESS in 58m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/62f8f7fd91554ff0be4ade4d983bbf8b : SUCCESS in 56m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2460792cd934473e9c7c2dda931c8db8 : SUCCESS in 1h 23m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b96a1e47d3b542a68c7bbae27c5282df : RETRY_LIMIT in 8m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/99058b5580da41f9919f6c4fbccb28a6 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f45f0f122e574152b23fabed0e2ac5bf : SUCCESS in 21m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aea2b7b73daa417fa6dbc54a4bc80bbc : SUCCESS in 1h 00m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2ad29bd42af4372b287f7bb9a6fee86 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f82e10d3ce94145af747b33fe99b7b8 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c486dba3495e4c00a9f8a413c00650a0 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb115798441f4c2cb2d1a382168a93b1 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5614a1b99aa46679f0e755eda8366f6 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bbd26215c384a7d8cbbe0bcd5280f7f : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7d9de5d625c4c0597918a989533332e : SUCCESS in 24m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f14c089aaba4093834ec295d54144da : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd370afaaf6644d1a537434dd333e1ab : SUCCESS in 1h 01m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d3a0252f3514281b6ee02df0264a0fd : SUCCESS in 1h 01m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ad0d8519faa41c1bef0c84c5bf6ab04 : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":1},{"id":"24dc18680ff7269d7945263e70729b96ebded42b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-30 09:14:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b07dd318714acc0f19135112b2897f8194af6493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 09:15:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2440b4ad95fe8f1696853d57158fc154b44c71c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 09:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/38b423ed440946639bfb5aea05dc8099 : SUCCESS in 1h 00m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fb4603df9b94fb5bc0fdd5d2f2e38b3 : SUCCESS in 43m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf2ce4ec5b524441a77f123fc4d501cc : SUCCESS in 1h 39m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0beaca37ff341b49ae61c3fa1d30b6c : SUCCESS in 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24013dbd58e5b6e0671b301025d33a5c9aea8295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:34:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/226153d84b944b1baca353c39ac13e5c : SUCCESS in 38m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ef061875d2184fefa71017cf0bc37b83 : SUCCESS in 1h 15m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b164aa045b9942dd924401b248cb8dd3 : SUCCESS in 1h 01m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37d72bc0b49b48108d94941e9b25d577 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae531d4663e465891620b3962c614c1 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b5f056d3e84461db1c6948cbddc787e : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56bd2ab6206f438bb3b21f8ae12a9ceb : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea71bf3ad2849949d3431c31a0da7a4 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b98608a826bf43ffb62f7cc913bb81c9 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8818b29af2654db19017279e3f0ad89c : SUCCESS in 35m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d66d5462f62a404fba627ac1fbd6ed3b : SUCCESS in 1h 18m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6bd0a0d743e41eb8bff738e7c2a3b41 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"56e82d231181b5dbff82c076a4021f1850cf8325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:34:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"375d365f3857bc5d4c15d0f67abcde6dadcf8ced","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:36:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d859f9d8184df19886fb546916ad82 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42c1c5c8794d474497f43a6b50018c3e : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a221d08ae3e48f9d0a50de477ec4f876da42fcdc","revisions":{"a221d08ae3e48f9d0a50de477ec4f876da42fcdc":{"kind":"REWORK","_number":1,"created":"2020-06-25 09:23:05.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/34/737934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/737934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/737934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/737934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/737934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/737934/1"}}},"commit":{"parents":[{"commit":"4103b0b151af100493c7242b088b42f051bf5485","subject":"Fix horizon debian binary docker image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4103b0b151af100493c7242b088b42f051bf5485"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-25 08:41:45.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-25 09:21:42.000000000","tz":60},"subject":"Fix Monasca Agent Statsd service under Py3","message":"Fix Monasca Agent Statsd service under Py3\n\nPulls in this fix:\nhttps://review.opendev.org/#/c/737929/\n\nDepends-On: https://review.opendev.org/#/c/737933/\nCloses-Bug: #1885104\nChange-Id: I014c9a52b2f61a86e3a8adafdf4783c3f58148ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a221d08ae3e48f9d0a50de477ec4f876da42fcdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a221d08ae3e48f9d0a50de477ec4f876da42fcdc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
