)]}'
{"id":"openstack%2Fkolla~738256","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I9f4a21bbf1c8383e21e59350206aba3424602270","project":"openstack/kolla","branch":"stable/train","topic":"bug/1883233-stable/train","hashtags":[],"change_id":"I9f4a21bbf1c8383e21e59350206aba3424602270","subject":"Change lang package block order","status":"MERGED","created":"2020-06-26 18:08:02.000000000","updated":"2020-06-26 22:53:04.000000000","submitted":"2020-06-26 22:51:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738256-1593211902977-0147cf3e","meta_rev_id":"37cab0c1320e22a107b3b39b573a6cccc8144afe","_number":738256,"virtual_id_number":738256,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 22:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},{"value":2,"date":"2020-06-26 21:05:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 18:08:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},{"value":1,"date":"2020-06-26 21:05:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-06-26 18:08:02.000000000","_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-06-26 18:08:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 18:08:02.000000000","updated_by":{"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},"reviewer":{"_account_id":31763,"name":"Gaël THEROND","username":"gtherond","inactive":true},"state":"REVIEWER"},{"updated":"2020-06-26 18:08:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-26 21:05:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-26 22:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e12f3b0c36e78f1a809fe7d069affcfc93de13bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 18:08:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1b82639e491e097e529cef25ea68c8535c931bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 18:08:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09730377e0899c07b6e2370d7c934045c7476185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0f7b2221aa664b06b298c03ec62d36de : SUCCESS in 42m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/233ac79c76744e39877e92ad2b72aaf5 : SUCCESS in 52m 59s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/92e2d30f602f4103a9ed3514f7e0522f : SUCCESS in 27m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/93f49e2d16574d77bd02882a87f2b8ee : SUCCESS in 32m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b22dfe2de18544e79d4227e035404c17 : SUCCESS in 1h 45m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/64b49e9a0c4e44739b8350d8d75f8dc5 : SUCCESS in 1h 17m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cb4b683cb08c49b98b6efed4ec092b08 : SUCCESS in 1h 16m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b46207d0eb0a4ea6bbe16bcb68a8d438 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f0851a3b300e4579b0a86503824f1c7a : SUCCESS in 1h 42m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/80d08caaa59b4ca3b2241d90d5e9d3e1 : SUCCESS in 53m 14s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/4c658a1612be46b8b46f75d63f7d2da6 : SUCCESS in 39m 51s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/26446f0da63847be8f2ac904b22c5628 : SUCCESS in 31m 55s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/180ebf86f8b740e19e19a9b6c645c6c9 : SUCCESS in 2h 16m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e2f492788720449188ac6bd212172ce7 : FAILURE in 1h 04m 27s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/2d70abdafb114e05b6ebbe1552f1cc60 : TIMED_OUT in 2h 03m 17s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/17c96391a99c42a2b87e29c3545febea : SUCCESS in 1h 50m 30s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5a25f6836df9435fad5d525c509cd003 : SUCCESS in 1h 53m 10s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e46bf97c14ec415f838b3e1c808f859a : FAILURE in 1h 08m 40s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/2f66ee8de191410d891bdf2c23907bd9 : SUCCESS in 1h 34m 14s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/4098612919554e7b930ffaa2d4af109c : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2e3b8402abc4ef4bfce679fe6172b55 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d97267c68faa410e876d889ee48d2a65 : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7da287dcba4847ba8ae7359ba440e14a : TIMED_OUT in 2h 02m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c33909a4dbf414b8fdb413353ee3e60 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4f1eadfb474cb29c651350799df81c : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1fa0663789f4d2eac36d5c932a4c386 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14987db17aba4494b729fb65d2aa5b66 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d65d696e0ab4573b7d84bbbee5be38c : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3dbb7fa75954f8d8a7dc56cb9b4b137 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57bf13928944461d97fc04096c6a6464 : SUCCESS in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14e8c04924984a6f8f7e4d6ac1163608 : SUCCESS in 24m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a9d4e4bd2d54f4c995f7d038d5059b2 : SUCCESS in 45m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0acb056efe84326855438f1020bdb6c : SUCCESS in 57m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14e71959475a4eb8a1105cd076bd8dcc : SUCCESS in 1h 14m 32s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d4751e345746482bb1286bc5b04c785d : SUCCESS in 1h 37m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/76f43efe8b7b4395a8af191f0c70a0e1 : SUCCESS in 1h 55m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8fd7471aa70f4c0d8b5b39eec738281d : SUCCESS in 1h 47m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d7cc231f91404869a446f127e89b30c2 : SUCCESS in 1h 04m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"007f31e9df07198e804d571c09e6739c303baa23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 21:05:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d17b53d039aede79685b9bca9d7b93b596d4bd0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:05:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b076e72c73102337d025f4639393cecf8d7e0cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 22:51:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/38112fe3b282423ba602465d0681d8be : SUCCESS in 46m 29s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/028bee10c97045cba6f5c1ca1272561b : SUCCESS in 32m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1d75b6bc836f44d481f1f859dfbe25e6 : SUCCESS in 1h 33m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4563c6b255794e3792938e983b446f6d : SUCCESS in 1h 22m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0dfa3f22535f4bb38726fee8e01e72b5 : SUCCESS in 1h 41m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4809ff5090874fd9afce0ccbecf36b9b : SUCCESS in 1h 02m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c85d283bab645ef84e3e27a33241b9d : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96091540f1d4b28b45a5ec36c8429ae : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73a2db4184424718b0a3eaa317677814 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1cddd25400f4a03bad63619d9cdad89 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb9e66f2eab4e87a3fd5bc6b5ec41b5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ad58d2907d4972b6cbf1949ca0819a : SUCCESS in 5m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c38dee9353a645aa9e05c9d8b1679f98 : SUCCESS in 26m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a6816cd42694cc49ccd45e37138c5b2 : SUCCESS in 53m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95bca3b91c9e4d1cad10f1b56598be17 : SUCCESS in 1h 13m 10s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"182503b38f1c494a07c7819ec82ac6cb2e90bee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 22:51:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"846329bfaacaac6637d018b879bb07ef6ba0632a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 22:53:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281c927aa2b54a66800d4bda36b919bf : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97d74c37b62223a9f61a9b0fdb5e316d18fb4fab","revisions":{"97d74c37b62223a9f61a9b0fdb5e316d18fb4fab":{"kind":"REWORK","_number":1,"created":"2020-06-26 18:08:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/738256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/738256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/738256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/738256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/738256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/738256/1"}}},"commit":{"parents":[{"commit":"5b03c3660e3180494d7b2e10021f87ffe3e884ed","subject":"Merge \"Fix missing nfs-ganesha-rados-urls package for CentOS\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b03c3660e3180494d7b2e10021f87ffe3e884ed"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2020-06-12 14:22:22.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-26 18:07:52.000000000","tz":120},"subject":"Change lang package block order","message":"Change lang package block order\n\nFix offline installation of language pack packages.\nChange-Id: I9f4a21bbf1c8383e21e59350206aba3424602270\nCloses-bug: #1883233\n(cherry picked from commit f9def021d86caecfd57018f41792f8a7c6bc6a80)\n(cherry picked from commit 714e1feaea947b6a014d34e5a9cecfc777472ca1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d74c37b62223a9f61a9b0fdb5e316d18fb4fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d74c37b62223a9f61a9b0fdb5e316d18fb4fab"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
