)]}'
{"id":"openstack%2Fkolla~738906","triplet_id":"openstack%2Fkolla~master~Iff600179630484228fb359417687e3d2a1a8b9f9","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iff600179630484228fb359417687e3d2a1a8b9f9","subject":"version-check: handle using of stable branch tarballs","status":"MERGED","created":"2020-07-01 17:50:55.000000000","updated":"2020-08-31 08:51:25.000000000","submitted":"2020-08-28 17:59:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738906-1598637592934-98515791","meta_rev_id":"81892f9bd0578e8a7fac834fd8e7285041857bca","_number":738906,"virtual_id_number":738906,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 17:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2020-08-28 15:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-08-28 15:00:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-08-28 15:00:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-28 15:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-08-28 15:00:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-08-31 08:51:25.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 16:11:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-27 15:53:11.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-08-28 10:04:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-08-28 15:00:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-28 17:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"def11b410e2f33ec1dec192be04cc4852226e801","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-01 17:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6221fc058b6c14a95abae3f673a50d232e813db2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-01 18:10:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8a08a2e43e557d117f61855e66d549b899fa03d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 19:43:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0d522b1ef64942b7acdee7eb16547e77 : SUCCESS in 24m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fced4e4154ad423f8afe2c24461be84c : SUCCESS in 33m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4023b7e772f940de834d1cd5b3139498 : SUCCESS in 1h 24m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ea7d1bae30bb4ca395621d42c6d2dbb8 : SUCCESS in 1h 15m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3cc8b75d5f40427581c748a076d027f2 : SUCCESS in 1h 25m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/182b8219d49d4621a7253f98f7830c76 : SUCCESS in 28m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33be53e450264da8acd607d42ab35613 : SUCCESS in 23m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1675d3d77c4249db8969943fe4db96e9 : SUCCESS in 1h 04m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a937bb26104e4c86b9bc0afe60c719c5 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd4d1eb224e4e0a8d2a8a82fe39d708 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76df0e140ac84b088d3594f0edec53b2 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43140f5faa94f60a436d56900d88e69 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e865b6e9e034439a99132a0a1ad2e5a : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a2766354c174564aaef2f2c0bba33f5 : SUCCESS in 27m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/287216641af74b96b5e8a3e3ad1e17c7 : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/00181688d5f6407fa9a7687074ccb8d5 : SUCCESS in 1h 02m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a4abd0516994f06b7d5600977a8582c : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e0378646684452896c45a3188e028ef : SUCCESS in 1h 09m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4dfd93bf295c2fe7739e5a07f537ec71f96a534a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 20:12:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9471930d23546249106f1ecfd0a3099 : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e00572e74c4643a9b8fdd2c32e233c1c : SUCCESS in 1h 08m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e86fc6a0e0b249fb9da541628f191c48 : SUCCESS in 1h 52m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/30fdf891a850463eb679fd29ad8ca546 : SUCCESS in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e15129f78f4805477ba266817234b8701e41423f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 11:08:20.000000000","message":"Patch Set 2:\n\nIn master we have a config that uses $openstack_branch. Can we not keep that?","accounts_in_message":[],"_revision_number":2},{"id":"83d6e3f20741d1ee52f510942ede6fae4618d566","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-02 11:26:16.000000000","message":"Patch Set 2:\n\n\u003e In master we have a config that uses $openstack_branch. Can we not\n\u003e keep that?\n\nMaster one was done by hand and does not handle situation when there are no branch tarballs available.\n\nNote STALE_PROJECTS list in patch. It is for all situation where we get releases done from stable/* but no stable branch tarballs. It does not matter in master but fails in stable/*.","accounts_in_message":[],"_revision_number":2},{"id":"8cc15c54b68895ae276ac2be0311a39f18e00bf1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 09:30:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"521fb9e3b3ed462680470841b14574fe78c64341","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-06 09:47:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ccd505a893ab6d6677a488ba9b074e3e2c99c652","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-06 09:48:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1037579633001ac5d622a3af5134b7ddd0519055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 10:51:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/71c08d456045484994544f931b2f6060 : FAILURE in 58m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/281ea3ebca3a43faab7a522e879c6bf7 : FAILURE in 1h 03m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1ba8ccf27eb4788bf0984f03d43d630 : FAILURE in 58m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0413c8b28bcd4ba3a185013b6679572f : FAILURE in 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6cf5b007879bd0cb7fec5e67535b3ab37798625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 11:23:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59cc2dd6ce254e98b0f24467f7eab31e : SUCCESS in 22m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68e9f9f884e14a95b6d26bfce60907c1 : SUCCESS in 22m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/feac1f2b1e72470b970e46532ec12b43 : SUCCESS in 1h 11m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/27993770ea3243bbbdbbc5b53603c106 : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6304a7aeaefc41c882a7933c2f0965f4 : SUCCESS in 1h 24m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/79358e39a7fd41919e98b025231afcda : SUCCESS in 22m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cffa233f505e4a5594157d05c3ca4f93 : SUCCESS in 32m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/17ca194f80a74c3991826b35290cd149 : SUCCESS in 1h 06m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68dbca694a7e4222b1872aadefb67300 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e4cff819904bcab67f5beb6de7a1a5 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec4e08a402f348eabfc14d1ce051e15b : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/632821e8f8764171969efae272f3c8ae : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a4e3a470034fd483d4dfc253d3650f : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/306e0f86dacc4e8cb46a53f73f0d610a : SUCCESS in 19m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc91c9b6aeb94c878c36513c5c9081ec : SUCCESS in 48m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b616261d983445cb6b86156de6c46a4 : SUCCESS in 1h 18m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2bc8e696fd284dee9c43a591187e5bb1 : SUCCESS in 1h 32m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/78c3f62013c74711b2d61a5e36466dc7 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":3},{"id":"e2b06343335ef466767310864cb8503ed435d046","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 13:03:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c70ea31f9d4202ba3d93f0d06f1a476277827c08","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-07 15:57:59.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a058e363cf63b9e811aa024079ffa738c843b739","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-10 08:09:16.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nfine otherwise","accounts_in_message":[],"_revision_number":3},{"id":"7e616622ba839f466f1c9d261991b2a94f6e8abe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 10:36:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c83c31a10dabf81fec32217686a46c9bfb2de88","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 10:38:53.000000000","message":"Patch Set 3:\n\nWould be nice to get this one merged","accounts_in_message":[],"_revision_number":3},{"id":"393aba220c2cb48df02964c61d2caf971a0ea040","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-26 14:58:30.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"9a54d7141fffc29b486125c9921727081b06d592","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-26 16:09:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b19eb7bc38525d16c4a5c9ddedc0022f00133ba8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-26 16:11:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9f1b3110b84f519c418ddd268b235565b2bdbce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:50:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ffafbfa132434734a0e69e62b490bca3 : SUCCESS in 21m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa95d3d8c5ea4e878fc723f534139fb8 : FAILURE in 22m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/32301e53ab6e456a8d0621b8253dc296 : SUCCESS in 1h 08m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e04acc89ec604b6289645b4d41615599 : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b792143e5ca349c3809f4abd92bf51ce : SUCCESS in 1h 12m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5bf7b97bcb2b4b87a572b5e298f20b9b : SUCCESS in 38m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ad8bca231b7e44c081bd8ea3f75eaf12 : SUCCESS in 14m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a66008356a3b40419ac642c972b615ba : SUCCESS in 58m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42ddd46c3b6b4a61a44d4fd1e3bd5dec : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c97da163b8346348a07f6778efb115f : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81065b75bbc246c18d4ad3ef9c8ed0f3 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fc0155dc532442fbc2798ae9f699abe : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8f9208c9a74e4d9d141a0b561df0c5 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f020a65d869d48b8ab798eeab9e0903a : SUCCESS in 22m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7413ca1e3faa432ebe4d0461db613f78 : FAILURE in 1h 14m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17f66584c3d74ffa992aa32e68e4b261 : SUCCESS in 52m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14eaeb98ee0a41f69ef98301a5f381d4 : SUCCESS in 55m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8162e85a634c4db98cca00ecd24e458a : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":4},{"id":"32a418ebd3f743e83977f447fc528fda83845885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:51:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5147173d4f84616a6e5405e93db8d63 : FAILURE in 38m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8b5520ed87041b3952d0547799c5d34 : SUCCESS in 30m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f010f717eb8046c09a0e56c947bfb339 : SUCCESS in 1h 22m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b39bdcad48ec4061a46c8dd1baa7aeb4 : FAILURE in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1eb278e7bcfe043e2271f338d897257ebca6ce78","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-27 15:53:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"991d7a1ebd428391362dd16c5befdd33d005c6da","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-28 10:04:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4d54b813c84a43fe57bdeb779bdbd2d668a823f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 15:00:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"193e1ac62029606d2568ffd732358e7d82d2a712","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 15:00:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1f6f27de4a304c0aaabd9ff0e970155bb78d479f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 15:00:57.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c606aba906f36a74b152944b366706bc9554f048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:28:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/34ef988517cd4a668d1892d2f3d81b25 : SUCCESS in 50m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3ac4224ea9f943d7984f561e37602450 : SUCCESS in 51m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d8cc42d54f2b406fb8df70f637c1c821 : SUCCESS in 1h 14m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/90126b25a355428aba0ebce26788cbd5 : SUCCESS in 1h 20m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/56f2339a03c04f9ba9de709d143bf925 : SUCCESS in 1h 06m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee529b3d86f348fda37e79d3621bdaca : SUCCESS in 19m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b78c7e3bdff4e8f87444111399ea74f : SUCCESS in 16m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/642176137ddd448ca4eba71bfa571c0d : SUCCESS in 1h 17m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67f79656b5c34fd0889898ab24ee218b : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982538044f794ccfba888570348590d1 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d419f4454259402c8fd0c74e76268b68 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33bad5a85d4f406bac515cf1b9ae6481 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e93550fc97462889fa4eb8ee098223 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec8ec069216f4a33972f62df6ad4e977 : SUCCESS in 16m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2367ceb1b2f940878c81e744970437fe : SUCCESS in 22m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d9e745868242dfb192f34a1459f3b8 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f05b6d990c842d3aa419d77a646a77b : FAILURE in 48m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/caba7139d74144089ddad860f1296b07 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":5},{"id":"bd6a002f12e55abdb2dedada7b8625fc0fb84817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:29:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"24a130d43588ad516d7802ee420d2dab3438c0d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:43:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e40a1c1a56748aa87650a8a8cdc889e : SUCCESS in 52m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c76d83691db47019cf79d836ceef755 : SUCCESS in 48m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc6a53bc001e4d1ca815e6d8c72f5d73 : SUCCESS in 1h 36m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a4cba818f0443a084824f463533a475 : SUCCESS in 45m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50a8105507659adf71a96e7007fbea2639db1462","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:59:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a8cfb720d063b3831c55ed76e72a62cea83feef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:59:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8cd5b77ffd8f403aa143d285d6efe02e : SUCCESS in 25m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1ecc2b8c16f74b748b224d5bcacaf1f3 : SUCCESS in 1h 04m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b4520e3e16394f6584cf4da55f41471b : SUCCESS in 1h 25m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1365ee8641542509bda8e8db4745e0a : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffcb09bacfc44296a58885606ed0de62 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25f5f5a79f3d4f4fa365ce326e624fd9 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386adce835b94afeb0c8bbf13e901024 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e2bf33e94d4da8bd0b20a2a7e9f403 : SUCCESS in 8m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac97dd2c07c144cfb373d1faef3bbfe4 : SUCCESS in 20m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0dc367b14efc49de9094cbfe6a306f63 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8c45d1b36bf4941a2801dc5a9eddb6a : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":5},{"id":"703ecb650492451c91a1141811719db0e78d26ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 18:01:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd040a002de4abfb21f4686958c2058 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":5},{"id":"46e61186ba0ec0bf475dbc9611aa1178dd96eaf9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-31 08:51:25.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 863b958e849fb671ac2d3fa8582c458af66017b6","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2cfc25e53a8bb6cb21bee24051c65d36015c6c09","revisions":{"18efe80edf1b72a19567cc1ea091051c3a22c367":{"kind":"REWORK","_number":1,"created":"2020-07-01 17:50:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/06/738906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/738906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/738906/1"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18efe80edf1b72a19567cc1ea091051c3a22c367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18efe80edf1b72a19567cc1ea091051c3a22c367"}]},"branch":"refs/heads/master"},"8b845538c22062453a10318e60b8da163b353479":{"kind":"REWORK","_number":2,"created":"2020-07-01 18:10:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/06/738906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/738906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/738906/2"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 18:10:37.000000000","tz":0},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b845538c22062453a10318e60b8da163b353479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b845538c22062453a10318e60b8da163b353479"}]},"branch":"refs/heads/master"},"ff783354ef43ad2f0134f4afe4028a99af9b3b8c":{"kind":"REWORK","_number":3,"created":"2020-07-06 09:47:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/06/738906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/738906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/738906/3"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:47:28.000000000","tz":120},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff783354ef43ad2f0134f4afe4028a99af9b3b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff783354ef43ad2f0134f4afe4028a99af9b3b8c"}]},"branch":"refs/heads/master"},"703393624ecddb5a2dd556f406b040cb211e5b81":{"kind":"REWORK","_number":4,"created":"2020-08-26 16:09:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/738906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/738906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/738906/4"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-26 16:09:38.000000000","tz":0},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/703393624ecddb5a2dd556f406b040cb211e5b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/703393624ecddb5a2dd556f406b040cb211e5b81"}]},"branch":"refs/heads/master"},"2cfc25e53a8bb6cb21bee24051c65d36015c6c09":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-28 15:00:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/06/738906/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/738906/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/738906/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/738906/5"}}},"commit":{"parents":[{"commit":"b4dde581c528c2072e408768601b75a97551e5a8","subject":"Merge \"bifrost: stop sourcing env-vars\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4dde581c528c2072e408768601b75a97551e5a8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:50:22.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-28 15:00:55.000000000","tz":0},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cfc25e53a8bb6cb21bee24051c65d36015c6c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cfc25e53a8bb6cb21bee24051c65d36015c6c09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
