)]}'
{"id":"openstack%2Fkolla~738907","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I0d8fe379d27962b5adf604d1c999371d52f570ed","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I0d8fe379d27962b5adf604d1c999371d52f570ed","subject":"version-check: handle using of stable branch tarballs","status":"ABANDONED","created":"2020-07-01 17:51:17.000000000","updated":"2020-07-08 15:11:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9be3c83dfaf1fee6c2b4c8a5848e6139fb27191a","_number":738907,"virtual_id_number":738907,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-02 15:08:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-02 16:59:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2020-07-08 15:10:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 16:59:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 15:10:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"268fc167ab6b961aee7c4d47d7bea81cf43a3a6e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-01 17:51:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c3037191c44f0d89bb69e64f301e02eb21eaa45","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-01 18:11:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"49200afcdf067f2403bd4eb8930455d070d09f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 20:20:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/177d87506c1b41a5b3feb048d23bef85 : SUCCESS in 21m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/110ffa5154ee4cdeb7342b3591e4ae21 : SUCCESS in 1h 51m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/413017683b8d4cffacc6b3cc6c73948c : TIMED_OUT in 2h 06m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/33c76ef9204c4503b7adcd30f11c2c6c : SUCCESS in 1h 44m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f1576b23e640458797f636834f2df6b2 : SUCCESS in 1h 53m 35s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3c85b138529d4fde89c0b01704f5787e : SUCCESS in 46m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8d52530684094fd883069dedf09e0087 : SUCCESS in 1h 42m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e3cb310908f43faa0aef4ad54960b58 : SUCCESS in 17m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/363c04e246594f94af98c36b4ed069db : SUCCESS in 1h 56m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4feef07c0424bf09fbd0611153df237 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffffa7f61d8f4f05b044aeaeaf40f9ad : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34f00e5cf9c34e89a430aa1775dfd1fa : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eef59522571047e39851576d4e3cbabe : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37f7670e78f4c56ab3fbe0426ddc023 : SUCCESS in 13m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2fe162c729e40f8b0ab447cd786272c : SUCCESS in 1h 32m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c312ab346284e2ea7b4213142b45f68 : SUCCESS in 1h 34m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ca0b5f1685b4f96b0852dd0a09e0cf4 : SUCCESS in 1h 58m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/420e491851fa4cd198bae02801248e79 : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee519302b42f4ce781c01a82bb8efe23 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"fde34adbd732c5676e954ba8306f9f8fd77b08bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 20:22:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebf272a8586a40fbb31e8a7d18ac7024 : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c217c60c2a7e43e39da3822beaf127cd : SUCCESS in 1h 12m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/25c786fefaf54f10b21b2d7c4d428382 : SUCCESS in 2h 03m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8188c3fc12a4ae1a391fdb0d76cf21f : SUCCESS in 1h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1d96087ee5dc8d556d2cd8fa618edaf0fb250de","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-02 15:08:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"05af6622d69201af55f63e2d41d970368a73c1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 16:59:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/076fa1050670438599b9c7f1ae10c5cf : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/32a5d5a34c244dae821ee4f78c3b5876 : SUCCESS in 28m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6c3addfbb9d7458196f9ed89eedfece0 : SUCCESS in 1h 22m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/92523d833598429ea5198dd2fe6b6eae : SUCCESS in 1h 08m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d055a81e61b84f4492cd1c283ecbeaec : SUCCESS in 1h 25m 31s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/31e153564d96416194ad78496a31d75c : SUCCESS in 41m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/176953c52d924aba9c6cd688ba9bebb2 : SUCCESS in 21m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6dfb0ef1f044fd391e2655167c0ec4c : SUCCESS in 27m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b0c2ae7dbd144bf59126a002bf8e2eaf : SUCCESS in 1h 15m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b27ad417da5c46e0a5807bad1e57ab68 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894b4dd560c14e4d92db31c81551bc6d : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a717a46c23489fba52925195429990 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ecdfaffbc6d4fa29e47ae021018fdd5 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d04b77e9fd432197f78416404b4f2d : SUCCESS in 7m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d76a83b4e0cd4c67bad3f10e23b4c7d3 : SUCCESS in 30m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a67485c5b7f46cea9dd55297d00c4b7 : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17f1ceb8466443739191dbc0d06820aa : SUCCESS in 1h 46m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a00087c2e93d4a5dbc69a480425e51c1 : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cdc7872e041b46efb68e493e99c0085b : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f305d58c88d9efb708b825ba3adf03a511abfcb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b00b22d6c120485c881c70c9dc3e946a : SUCCESS in 1h 29m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc732b56225c4cdeb7d07e1280ca8448 : SUCCESS in 1h 22m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c9ce9c0e09740d283db789df42ce410 : SUCCESS in 2h 40m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bee4082625d545c195a270eaa217d27d : SUCCESS in 1h 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b6f17076241d7988b966871ebd547f7d81a52231","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-08 15:10:42.000000000","message":"Patch Set 2: Code-Review-1\n\nplease backport properly when master merges, thanks","accounts_in_message":[],"_revision_number":2},{"id":"9be3c83dfaf1fee6c2b4c8a5848e6139fb27191a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-08 15:11:31.000000000","message":"Abandoned\n\nThis was CI test. worked fine.\n\nwill return as proper backport","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b07c31ef7674110d589e781554300597f2f21058","revisions":{"831bcd0bcd5621f99b14e6ec4a2e3e727591e895":{"kind":"REWORK","_number":1,"created":"2020-07-01 17:51:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/738907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/738907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/738907/1"}}},"commit":{"parents":[{"commit":"58b764729a038e9415081316133e0fa4c8cd0979","subject":"Merge \"Fix Monasca Agent Statsd service under Py3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58b764729a038e9415081316133e0fa4c8cd0979"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:47:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:47:22.000000000","tz":120},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: I0d8fe379d27962b5adf604d1c999371d52f570ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/831bcd0bcd5621f99b14e6ec4a2e3e727591e895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/831bcd0bcd5621f99b14e6ec4a2e3e727591e895"}]},"branch":"refs/heads/stable/ussuri"},"b07c31ef7674110d589e781554300597f2f21058":{"kind":"REWORK","_number":2,"created":"2020-07-01 18:11:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/738907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/738907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/738907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/738907/2"}}},"commit":{"parents":[{"commit":"58b764729a038e9415081316133e0fa4c8cd0979","subject":"Merge \"Fix Monasca Agent Statsd service under Py3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58b764729a038e9415081316133e0fa4c8cd0979"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 17:47:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-01 18:11:52.000000000","tz":0},"subject":"version-check: handle using of stable branch tarballs","message":"version-check: handle using of stable branch tarballs\n\nFor Ussuri we decided to follow stable branch tarballs instead of using\nrelease ones. This gives us all fixes just as they land in a repository.\n\nChange-Id: Iff600179630484228fb359417687e3d2a1a8b9f9\n(cherry-picked from master + updated SOURCES in config.py)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07c31ef7674110d589e781554300597f2f21058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07c31ef7674110d589e781554300597f2f21058"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
