)]}'
{"id":"openstack%2Fkolla~738998","triplet_id":"openstack%2Fkolla~master~I0d8695fde2a70509c7535d0d582136c0b0e7667b","project":"openstack/kolla","branch":"master","topic":"remove-glance-registry","hashtags":[],"change_id":"I0d8695fde2a70509c7535d0d582136c0b0e7667b","subject":"Remove glance-registry container","status":"MERGED","created":"2020-07-02 09:13:49.000000000","updated":"2020-07-02 20:24:58.000000000","submitted":"2020-07-02 20:22:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738998-1593721358970-b4231495","meta_rev_id":"6a9948e5bf4a4588e94c7c9e362814ae3d4cd4b3","_number":738998,"virtual_id_number":738998,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-02 10:59:46.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 20:22:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-02 18:36:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2020-07-02 10:55:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-07-02 12:40:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-07-02 12:40:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 10:55:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-02 12:40:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-02 18:36:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-02 20:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec022102f614231aec96491afea6030911460524","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-02 09:13:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e5cd86b8787cd60336201fb5b41e45bfd506724","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-02 10:23:23.000000000","message":"Topic set to remove-glance-registry","accounts_in_message":[],"_revision_number":1},{"id":"a85cd0510b2b0438d32442faafd965ec60ee247c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 10:54:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce0b14344198896c2a13a5db7ba128a3444d1cbd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 10:54:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"75ad98cb0dcee13fbd9ecd70325fc843865130ab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 10:55:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4cba17cd75b20f1729f6ebb7b5e6bbb02711045e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 10:55:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b18eb4600fd36b4e84c7b9686f0218bec542133","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-02 10:59:46.000000000","message":"Patch Set 3:\n\nThanks for fixing it!","accounts_in_message":[],"_revision_number":3},{"id":"a786cedb0cfb42004b73176a87ba10d9f43bf717","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:23:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/73fd7184f1fb43389d13b46a188a2051 : SUCCESS in 48m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/405a1dc1e6fe472393e6afa9d7fcd36a : SUCCESS in 34m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55d59d0b3ae645a0a2ce766d692f3ca6 : SUCCESS in 1h 25m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dd7b962e3114dd3ba96cefff5a7c379 : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7fb6e92b4c07718669f7fb414ba77eb2a74de0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-02 12:40:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nfinally gone","accounts_in_message":[],"_revision_number":3},{"id":"63f18c5bcdb1574959d68ae5f2179a40419343d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:54:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7dfd88a70d35447985a5e4ce503e729f : SUCCESS in 23m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/92d6432afe734d2ba14eecb8a5ce2976 : SUCCESS in 1h 30m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f0d327f8c8764e08b09eb8deffbef1f4 : SUCCESS in 1h 15m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d537504332a54e86bad700383607281c : SUCCESS in 1h 10m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0bb3b1845a014ee1be0ee51a1c1a37f8 : SUCCESS in 1h 20m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f72f2b314b4f4ac1b8a578a8346644e7 : SUCCESS in 34m 39s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6ae186915f594b43aa22dce0963afe44 : SUCCESS in 1h 45m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2e508f24e3234eb7be14bbdb9111251f : SUCCESS in 1h 34m 54s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/c218fb9177b245aa8674ded81012991a : SUCCESS in 1h 51m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5e7ef9c1e424a02b0969df7b5a95c2a : SUCCESS in 31m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb678908c18b46e9b7d2b9d94054a246 : SUCCESS in 20m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f3ed89d667dc48c08a1e4419f4cf7992 : SUCCESS in 1h 29m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/936cd51be30d4bfa854ffa9790a6e26f : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752d8ee1805f4cf18348e4da9246ddce : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84ff23f896454e00ba1fe307167f0cae : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87fa4bd924b541648f527af1f637182a : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8cd52b041b49508f6b5ab1d5510da5 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b04bc7bba14f44a5c2089f15cc1d55 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/564522b23b6c4ccca1fcc7e46efffa4c : SUCCESS in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/013c5df97a27485e935f2f1d737d9e70 : SUCCESS in 24m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4fcd8c061fc436fb489eb61154e8615 : SUCCESS in 56m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66219a1a300c46c6ad23bf31f5e6f097 : SUCCESS in 1h 01m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/824f54c887cc4b4faf89d4b226779c23 : SUCCESS in 1h 15m 15s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e57a43baefd244158e5fc11d94dcd67d : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/959b6d95518b4e5792bfa47b1ee949a1 : SUCCESS in 1h 29m 05s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c871a80b865d487c95a9d3b6005642f6 : SUCCESS in 1h 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8dba015ae88bb18c633a57a9b71126d72e06258f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:55:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6902409015ca32ae66621c2198438fbefe461486","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-02 14:10:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c0de07cecd1f48d9b58408c83de6a27f : SUCCESS in 26m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73b0becab1a6713fd2332feb5a7e6c2b535c41e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 14:24:36.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a08f3dfbbe234ae49be1fa0e63b28a0c : SUCCESS in 36m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d11c5e09d3d94cb9a70ba05ef62199b8 : SUCCESS in 1h 08m 46s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c4d0057a1b934acc8170909f16142950 : FAILURE in 34m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76633bbc66714fca873a54082ef7ddb9 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a61eb8e3f048fa85a3c7a8835eafe1 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6715217a1148c08bc7385880487096 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7d8af08eff04e21b7e5f495b6b762e3 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8939d575084523a250afc08e0011c4 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083ed8d01e7a43e692d8a64a4fb6a9e7 : SUCCESS in 7m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c0e1ddb005342649e1ac3f332be409f : SUCCESS in 24m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2240ca9961904b9090beee845323d6a2 : SUCCESS in 1h 09m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b13a0b0a04214539aacf771023924ffc : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":3},{"id":"36ab2f05b337b094177ec1fac31d99fc9b472a31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 15:54:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2c00014a6380285501c8cfc810ea44d7bb852f57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:08:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebaa8e1dbd5440a6af6cedea60c2a2b7 : SUCCESS in 1h 34m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e4896463b834807be403a2ecd2b21e5 : SUCCESS in 1h 10m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d459b8c4a63840e5b36247d438ad7688 : SUCCESS in 2h 08m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/676a064b8c24494fa743e4cb534859db : SUCCESS in 1h 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfc1166df63881446a5dc1516acfdc5f50fdcdde","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-02 18:36:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6734a0f407ef412d93b594b24b6800cd : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"19239a09482079bf5c7a4b2307bb84864672a6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:42:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eded2361bd8945368e876e7956780769 : SUCCESS in 20m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6cbb5b4142e74ed1a470a9658d8fdba7 : SUCCESS in 26m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c198a60c947c404ebb32e4b1c24d1e94 : SUCCESS in 1h 53m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7cc0a297485c4bcf967572b69e52ffe5 : TIMED_OUT in 2h 06m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/33ab0f6c06c9469983963c62c640139f : SUCCESS in 1h 19m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e967740490d7453e9c2af66f3122d361 : SUCCESS in 37m 48s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/16f0d2eabdbe4507bbebdebf2f04e43a : FAILURE in 1h 18m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a1edf20b28174b3f953656fa6571aba1 : SUCCESS in 2h 21m 34s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b36850711f9b41cf87f425ae31bba947 : SUCCESS in 2h 01m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e92aea45bc47449f9122091542d0b03d : SUCCESS in 19m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d848cdce95a4c2396d762a0b74b45cb : SUCCESS in 45m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5116845545fc49628a33acaec321a8d0 : SUCCESS in 1h 05m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fb7ad34fcf8464985399e0f15a95396 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf25002514a468ca071b3f6360cdf52 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90b216013044dc1bae1809f7c813a34 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84692a226f4c4403a83cd46f95ee34ba : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658a1347346b4344b310b7bfa596df72 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67ec2d83642b40608b7e0c36e680b6cc : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d67baa7208754ad98f7867dc89a9a2cc : SUCCESS in 34m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07802f200be64fd0986740e12f01f7dd : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3fd6f999aa1542bd952e9ecf4f1a2e6e : SUCCESS in 1h 03m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae90be3eeac2420887757db277fb7afd : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d05c304f9568474ca5502d89493b612a : SUCCESS in 1h 08m 15s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/66ca2547707f4e3caa15994e5877876f : SUCCESS in 1h 39m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/556d5cd52c9143d2ba0872882842c9e1 : POST_FAILURE in 1h 45m 25s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/06612d495a1b467b9f3b0a19ee9adc78 : SUCCESS in 1h 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31c9e9374baf05a2ac847bdde8d78a6d851abb02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:42:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59d8a16f8e98fc3a574f7e5ee26288433777e403","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:22:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f776dea533ab585be9adfb0626003eb6c3a804c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:22:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cb26313780554ac0a02d8487582094ab : SUCCESS in 30m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d9baa48f667c4430a38f5ebf2849cbfa : SUCCESS in 1h 15m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/28c1e8492ef049afa7304f8400184993 : SUCCESS in 1h 33m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07c7d90c3a2e4d7bbefa19e9a5ccb71d : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058f0dfc040d4c0b9a9573b0fb6eeb72 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6d6606c4c64eff899140bf70ac8ddc : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c18375d31e8749a7a0ed9e04bd95d615 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913114d76e884fd18becd39a75f898c9 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7c0eff2a407463da8ce79b6038ee4f9 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f04be9b7c045416fae4ffff14007692f : SUCCESS in 34m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/445bee034df04841ade7f1650043e652 : SUCCESS in 1h 08m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67ab3eaa6c5b423398fa78ee9f25c1d9 : SUCCESS in 1h 33m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6a9948e5bf4a4588e94c7c9e362814ae3d4cd4b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:24:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afed4c4f15e435f93883d0479cd478a : SUCCESS in 1m 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb3ae0f2d4984a5ea8f2b354d36559b6 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3a3c34c5975a289ca1cbd546db07ccc9a7f318c9","revisions":{"873275d84a0257acd91bca4c235876f2faedcb2e":{"kind":"REWORK","_number":1,"created":"2020-07-02 09:13:49.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/98/738998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/738998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/738998/1"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-07-02 09:12:02.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-07-02 09:13:32.000000000","tz":120},"subject":"Remove glance-registry container","message":"Remove glance-registry container\n\nglance-registry service was deprecated in Queens release and has been\nremoved in Victoria [1].\n\n[1] https://review.opendev.org/#/c/738671/\n\nChange-Id: I0d8695fde2a70509c7535d0d582136c0b0e7667b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/873275d84a0257acd91bca4c235876f2faedcb2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/873275d84a0257acd91bca4c235876f2faedcb2e"}]},"branch":"refs/heads/master"},"b9d02c8a44a8f7a9f6cff35b8d74694335e6ef5b":{"kind":"REWORK","_number":2,"created":"2020-07-02 10:54:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/738998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/738998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/738998/2"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-07-02 09:12:02.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-02 10:54:35.000000000","tz":0},"subject":"Remove glance-registry container","message":"Remove glance-registry container\n\nglance-registry service was deprecated in Queens release and has been\nremoved in Victoria [1].\n\n[1] https://review.opendev.org/#/c/738671/\n\nChange-Id: I0d8695fde2a70509c7535d0d582136c0b0e7667b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9d02c8a44a8f7a9f6cff35b8d74694335e6ef5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9d02c8a44a8f7a9f6cff35b8d74694335e6ef5b"}]},"branch":"refs/heads/master"},"3a3c34c5975a289ca1cbd546db07ccc9a7f318c9":{"kind":"REWORK","_number":3,"created":"2020-07-02 10:55:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/738998/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/738998/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/738998/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/738998/3"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-07-02 09:12:02.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-02 10:54:57.000000000","tz":0},"subject":"Remove glance-registry container","message":"Remove glance-registry container\n\nglance-registry service was deprecated in Queens release and has been\nremoved in Victoria [1].\n\n[1] https://review.opendev.org/#/c/738671/\n\nChange-Id: I0d8695fde2a70509c7535d0d582136c0b0e7667b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a3c34c5975a289ca1cbd546db07ccc9a7f318c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a3c34c5975a289ca1cbd546db07ccc9a7f318c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
