)]}'
{"id":"openstack%2Fkolla~739097","triplet_id":"openstack%2Fkolla~master~Ia2306507d6ef97f01ff520dc4222bc48c1eb9843","project":"openstack/kolla","branch":"master","topic":"openstack_exporter_bump","hashtags":[],"change_id":"Ia2306507d6ef97f01ff520dc4222bc48c1eb9843","subject":"Bump prometheus-openstack-exporter version to 1.1.0","status":"MERGED","created":"2020-07-02 16:35:35.000000000","updated":"2020-07-03 08:56:33.000000000","submitted":"2020-07-02 20:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739097-1593721360663-f3086ce0","meta_rev_id":"b7f96f6eceed1952778a4e39f7342c73178f4897","_number":739097,"virtual_id_number":739097,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 20:22:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-02 18:58:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-07-02 16:54:40.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-07-02 16:52:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-07-02 18:43:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-07-02 16:36:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 16:36:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-02 16:52:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-02 18:43:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-07-02 18:58:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-02 20:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0e40eb054a816fd4546e1c6345968e6bc72d423","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-07-02 16:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ecd03867c6a1e7e99cfde5b013d96eaed7a1403","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-02 16:52:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"198f5eaae62c7012a7e895478c0c1680dbf98e78","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-02 16:54:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10b98a1e7c46123ef48d86219f0eef0dad24aaec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:36:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f124ce240944855ac8f95f37335e3bd : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8c170a04920435aa9fc22a03d03ef4f : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/46e0ae8ca55e4af682f36dfea9a49963 : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/29d6f6cf72c3448c9426a14ee01272da : SUCCESS in 1h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f05b3756fb94bee74bd9bbe25782622734bf5f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/394ce209067c4f7494439e41c8912df9 : SUCCESS in 32m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/413f251754384d20b184a99a49de1244 : SUCCESS in 43m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e5acc89bf91d46ba98d1b0448861a3bd : SUCCESS in 1h 59m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1c034fcb163b460d8ebc2e41d3799e70 : SUCCESS in 1h 15m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c2050990ec1a4a75b0bed0160460f8b3 : SUCCESS in 1h 22m 04s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1c654cdd3221493c945d28b481340360 : SUCCESS in 46m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a150ab8341c347e6bdd080837d38ee54 : SUCCESS in 31m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2eb145e8717f4fc1bea1e681002a3b58 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe5218a6ae31488ab309fdd9a06ad293 : SUCCESS in 1h 02m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9c6047c4653440db280023ea9c28196 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8003ae74644076bc2714230979ca28 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca16d538611a4b0ab872220235b4e230 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f2797bbe6904ca88d9e85d550c20762 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac6518159fe44eb949f1779e1712300 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4aad37992a054386bfc2d7096f7e08ba : SUCCESS in 23m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b273cafb7a64e0fa190300baef8b906 : SUCCESS in 40m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1a64436f10f4ffd97465fed9212807d : SUCCESS in 1h 02m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94c53018971644d39876fd9ffef0a8ff : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a205c3c4efb34c339e88b30b78062068 : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"905e50f9d07652f188353d2d4fac6075eddb7423","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-02 18:43:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0462a79bb0dbfe7432170ed4f06f5003087f09e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 18:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0efbdda8749b3db85405b38a419f0a14f6eacc1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-02 18:58:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/61a66e5bcd2c42df9678be2d149702f6 : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9de23a090d1bd2cdf4ad334b8e8adce8bf912a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:22:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f637ae0381c85752edf5bbe4ecf8d0ccaa396ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:22:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3fa61a1bafb74d91a212269ab658a86c : SUCCESS in 58m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9647c2dbcbf943c697229a89fcfc2e57 : SUCCESS in 1h 15m 44s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5eecc65e2e8d4e2cbad1e524a4f59e7b : SUCCESS in 1h 31m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c650e4c72ace4600b54618a616d21d67 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915f05ba8eb54516a5f93c59ba0a53df : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/003da80c377548c590a53a70fb80d6db : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb5c0022401c47b49e13ac2a2d53ca40 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00727284c2634bc596785d95a87a9f09 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4883da5d15324f75a6120bd8cceaefd3 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3651df860be2429b99d0912bf12b3b10 : SUCCESS in 1h 03m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75a4c8013f8e4f118658531ad5afbbf4 : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1bc89e4ccc3ebf8c307616c05ce5c7319ac6a8fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:26:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c95684851c5449ba2d465b7c84fc4a2 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4b26c25b89b312f02bc8a6deff730adfc226732c","author":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"date":"2020-07-03 08:56:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit be5da6d2c9b595cfc49e2c744d1464120f24e99e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2841004652d74f27f270b4fba5a68e6c845e173c","revisions":{"2841004652d74f27f270b4fba5a68e6c845e173c":{"kind":"REWORK","_number":1,"created":"2020-07-02 16:35:35.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/97/739097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/739097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/739097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/739097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/739097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/739097/1"}}},"commit":{"parents":[{"commit":"1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d4d23cf0ea1dbfd2e6acde3b5292697610ca15d"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-07-02 16:35:11.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-07-02 16:35:33.000000000","tz":-240},"subject":"Bump prometheus-openstack-exporter version to 1.1.0","message":"Bump prometheus-openstack-exporter version to 1.1.0\n\nChange-Id: Ia2306507d6ef97f01ff520dc4222bc48c1eb9843\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2841004652d74f27f270b4fba5a68e6c845e173c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2841004652d74f27f270b4fba5a68e6c845e173c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
