)]}'
{"id":"openstack%2Fkolla~739183","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ia2306507d6ef97f01ff520dc4222bc48c1eb9843","project":"openstack/kolla","branch":"stable/ussuri","topic":"openstack_exporter_bump-stable/ussuri","hashtags":[],"change_id":"Ia2306507d6ef97f01ff520dc4222bc48c1eb9843","subject":"Bump prometheus-openstack-exporter version to 1.1.0","status":"MERGED","created":"2020-07-03 08:56:33.000000000","updated":"2020-07-06 13:07:46.000000000","submitted":"2020-07-06 13:05:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739183-1594040706356-ec7d04c8","meta_rev_id":"f3db198b2f8d5eaca763169f16f8dc12a7e1ed14","_number":739183,"virtual_id_number":739183,"owner":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 13:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":2,"date":"2020-07-06 09:17:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-07-03 08:57:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":1,"date":"2020-07-06 10:59:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-07-03 08:56:33.000000000","_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 08:56:33.000000000","updated_by":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2020-07-03 08:57:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-06 10:59:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-06 13:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b214329e97c9995f72168e9a163b6b5c44f13f16","author":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"date":"2020-07-03 08:56:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d9e9d16d839ae33a605c52c7d07863f38276a9f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-03 08:57:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8fb828089dfa067c682950bd5afc7388d960a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 11:22:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aa1a810e89c940358b4a7bd3d0b92ce4 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/236c33ab37114aa7bf8fa860a6691e4b : SUCCESS in 31m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1e1bd06e16344488b50729cdcb744049 : SUCCESS in 1h 16m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/30e6639ac8b94602abdc609cb41e2499 : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/377775a4bf46408c91902c1a5aeee01a : SUCCESS in 1h 28m 39s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f733d33673c24182942eedf430725881 : SUCCESS in 48m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a27d3de09304d55b8a63c2f83ffe2d2 : SUCCESS in 24m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6d9d1f4f888c4fb0b7f7e24dda583759 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/821d55dd3f25498d823b30f55148d5f0 : SUCCESS in 1h 07m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cd455eb51f948d58c997e3928831616 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3a14c3e88545cdb42fb4f904fd3e1c : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1e5f28c8c94d849410152fb6f7522f : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb06693768064e9c8f868101bd8022af : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10b489259124638b14c64901a6facad : SUCCESS in 7m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3cf5011837204fe398d6e787a7d70543 : SUCCESS in 2h 20m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e453ad3073e41e4bd5905589fdc9e32 : SUCCESS in 2h 19m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b49fc1784d5c46478d2eba5ccc201e65 : TIMED_OUT in 2h 05m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e28b39f459954b8283afff6dee22642f : TIMED_OUT in 2h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80f2ba00111f4f6dae4f332d3bd2292d : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5a06fea58d21152ecaaef3bf16ddb2241228855d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 11:26:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/62265b775eab4ae3af8d2a8e4b2417f0 : SUCCESS in 1h 21m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/836ea81be9d24199ae9e91e3e76ac2b1 : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d49f24c8914439b83cca3f595c735c3 : SUCCESS in 2h 25m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/53baab15da254140aa2c9293ee5b5d03 : SUCCESS in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0350ec34e4fea37a5c2217282e538286b94c2176","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 09:17:47.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cf3d40358091ea32e08a2d74b73ffff15e79d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 10:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f4d93e713ae9410d84c7c478be26e5ca : SUCCESS in 29m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2427c99c88344810a29be9401d4afbe9 : SUCCESS in 29m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/daa0f8f680e243a8b6a71162346d2542 : SUCCESS in 1h 26m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0a27a28d396f405fbdc36c5e6c176587 : SUCCESS in 1h 11m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/be02a13d5e0645229f9c5e7887978b36 : SUCCESS in 1h 33m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/139a35c8d47946f192e96382cd8369ca : SUCCESS in 43m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85f6fea6c618449998c41de90656126f : SUCCESS in 25m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b9f6351b2ef34fdb9c2092904f8ec11d : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/edb64df73adb4071ab4c1e74996ba8cb : SUCCESS in 1h 09m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b51fe20441bb425592f1e913dbecdaf2 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312e86e3959144f9aa1adc35e1378795 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ad2d14d69243d4a325cbb17ea59a48 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1860672010a34acfad7496403044d47a : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e980141afa47be9f8d128dc1988c51 : SUCCESS in 7m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18554094069a403a82e04aaeb638696c : SUCCESS in 46m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d42f10d48b154384a5eaf65a1679444e : SUCCESS in 36m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71258337d520440fb85bf859713e8ce4 : SUCCESS in 1h 15m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0902132ea4e459f87088a1a2f9fcc85 : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7372d4dabd2c4994a5f6bab7ce028710 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1293468c41c237c5f31375041a3d8a45e5bb7c5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 10:59:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b41c25b8b462d9b3d47a1d9a1a134be04aed220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 11:00:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9c273d928ebd71255833f57d097cda75929be56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8afa6345ca2451baa9762f6d37dd572 : TIMED_OUT in 3h 01m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5688d5180ce4b7fbe3d58745c6f2ac0 : TIMED_OUT in 3h 01m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/39bde3b66a2a425dba36dac52c1ee8e9 : TIMED_OUT in 3h 02m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/346fbc6703d7401ea3af2b057f9d8b88 : TIMED_OUT in 3h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91434f515a3ea58e96fe96d2ac3d613eb8d48948","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:05:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/45c61a68319c402f9f6ae800bc1c18a8 : SUCCESS in 23m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5275bcebc17d4df694bf286aa3062aa7 : SUCCESS in 1h 15m 22s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5abc329f301d43bab58174768f54c85b : SUCCESS in 1h 19m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e5248b27ef5402e86ba62ab3ff11637 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdbe1acffb548d9a3587eb5d796d76c : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ca4fce6965436db2d9f71adfa17d1d : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3bb13c04a23407e8fd62896b320dada : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99937b6970049e6aa3b3b9ffa032643 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/491b527c53514dbbabd76333fcc5a1fb : SUCCESS in 1h 17m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3752df36d71441b91f3deef63adcf6a : SUCCESS in 1h 44m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7868b97c9c6143d4a76199881af1a390 : SUCCESS in 1h 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cf1566b9d78e1ce698e0a415b278b2a88b551be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:05:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2c0ef8f204c2bc234a8acf841b1f49cecbf6172","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ff14b203664b0f81c17821f16d03d1 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be5da6d2c9b595cfc49e2c744d1464120f24e99e","revisions":{"be5da6d2c9b595cfc49e2c744d1464120f24e99e":{"kind":"REWORK","_number":1,"created":"2020-07-03 08:56:33.000000000","uploader":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"ref":"refs/changes/83/739183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/739183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/739183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/739183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/739183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/739183/1"}}},"commit":{"parents":[{"commit":"0463037070bb90271795b7051178b70933071911","subject":"Merge \"CentOS/AArch64: RabbitMQ is back!\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0463037070bb90271795b7051178b70933071911"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-07-02 16:35:11.000000000","tz":-240},"committer":{"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","date":"2020-07-03 08:56:33.000000000","tz":0},"subject":"Bump prometheus-openstack-exporter version to 1.1.0","message":"Bump prometheus-openstack-exporter version to 1.1.0\n\nChange-Id: Ia2306507d6ef97f01ff520dc4222bc48c1eb9843\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n(cherry picked from commit 2841004652d74f27f270b4fba5a68e6c845e173c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be5da6d2c9b595cfc49e2c744d1464120f24e99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be5da6d2c9b595cfc49e2c744d1464120f24e99e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
