)]}'
{"id":"openstack%2Fkolla~739440","triplet_id":"openstack%2Fkolla~master~I0cfde46c29eedce77c70c1c677220814a801ff3a","project":"openstack/kolla","branch":"master","topic":"remove_sensu","hashtags":[],"change_id":"I0cfde46c29eedce77c70c1c677220814a801ff3a","subject":"Remove sensu images","status":"MERGED","created":"2020-07-06 09:32:23.000000000","updated":"2020-07-07 08:52:49.000000000","submitted":"2020-07-07 08:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739440-1594111848550-447c4ef9","meta_rev_id":"3dbd0b18c29740ba4b6772f48e1d710d4648c446","_number":739440,"virtual_id_number":739440,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-07 08:50:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 10:18:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-07-07 07:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-07-06 12:53:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-07-07 07:06:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 09:35:40.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-07-06 10:18:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 12:53:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-07 08:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26b973274166b438735c41e2226bf731968e48f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-06 09:32:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02e6057edb67751a452e81c8f5bdccdcde59649c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-06 09:35:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8f41ef25ddb6549cebd14a81e634bc732c184c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 09:47:14.000000000","message":"Patch Set 2:\n\nrelease note...","accounts_in_message":[],"_revision_number":2},{"id":"7243d85d6e439b8a2360768f32050ac02b7a2e83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 09:47:20.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"8340516fdc2ad2c8c68997ef0e64db2aaec0ec23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-06 09:50:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0a46781ebfa7008e0fde9a3b9b4ca4e8def2aa0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-06 10:18:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/44c15700b5f4442189485e47be3aff43 : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"081f9ae37a08e6e6d5686904193ffaab0a81d8e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 10:51:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2eaaacd3ef3842f48daa5d5db7e71384 : FAILURE in 52m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e3872d8f0f44325999a07a394a73df1 : FAILURE in 51m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66a32e21ba4645fd92911148fb1181da : FAILURE in 52m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/27708c5dbd904a37958fdf28656b16d4 : FAILURE in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"389b39223c06d6f8a7b6cf3345d8ddc31aff7f89","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 11:01:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"404a7653105f285d566dd67ddaeb1aede6038ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 11:57:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f1ee8a59e81147b8bc171b4929f29ef8 : SUCCESS in 25m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b70ce22742f3492c8e2b4985f678ed08 : SUCCESS in 27m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1db2b79a60c740e093c0577d42210f56 : SUCCESS in 1h 19m 18s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b0323e22d2a346c3bf2195ad8b9f6bed : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/88f68b0a05804749856536ace0d8e962 : SUCCESS in 1h 25m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/49b9b42b238e4b04a5675a63d8519c01 : SUCCESS in 33m 34s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/7dca57a9cf5d436f836e6904050bfe1d : SUCCESS in 36m 19s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/66b215551cf7469397c088a7159ac3e6 : SUCCESS in 1h 51m 17s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bf92bb4c0b994cc6a5c0d96e5f5eb602 : SUCCESS in 2h 02m 47s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/03245869275341ccacf91cbdcc6c3e7b : SUCCESS in 2h 02m 05s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3891863c16a140f2b45097bdc97cfe45 : SUCCESS in 1h 47m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/1802560b3af04263884fb02a26ed926c : SUCCESS in 1h 28m 21s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/5836f26a69364179a011959474a05dcc : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b52d8bced1bb459698f222ebe67667b6 : SUCCESS in 20m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f11bfdc3ccab4a92ae75653c97660c22 : SUCCESS in 20m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/99147668a7ef4c02bb8e4db0a7b0db03 : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d60847b7b31146cd949aed99bbe69591 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32681c6a5dd4f4b94007dc45087a342 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a78962bdf2143fa8097f9da2ff75ea5 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cdab42aff6743f9b809958d443760c2 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5f79da0b58484796df58cb5732c3b5 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e3a5b887d94fbc8cdb850a1bacba44 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b447df5ccdaa433595474d3adf91efd5 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e409e5ba9afb4e459038c647aa547d77 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d476cb08b8747378841d217e0dd042a : SUCCESS in 1h 04m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9db3ac2c0d344d7876bd0a89d729ad6 : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32cd73f2eddf4972a20398e7745b04be : SUCCESS in 1h 08m 14s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/881fcfbdf671473388cc6ec3a4091d3f : SUCCESS in 1h 27m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2dec577ced324b85b11060d30833f2ed : POST_FAILURE in 1h 39m 18s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ac7c0e71e9d34807ac439412a8da41a9 : SUCCESS in 1h 44m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f69c0f54409f464ca4a8a708c0745ebd : SUCCESS in 1h 11m 11s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8c892707bf8c4b0d985bc13d518a0398 : SUCCESS in 1h 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e1eac6ea4412553f63a25f6933e0955990998d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 12:53:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"05d2cdef8a8dd6ba6dbac0f51bae4e02d059314a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-07 07:06:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fbdf98967d0ae4d358efc7a682c078138aa8d7ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 07:06:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60cdce4ad365876ad49d717849bec0e9d12b2aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:50:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6afd0358529aa7fd915138a2eb3ae9ea0237c099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:50:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad8656cbf46e4810a3c48a58717fae80 : SUCCESS in 39m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5c87be633ba74f7f84e04715b508d141 : SUCCESS in 1h 18m 44s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2d25a1cc58574910a41d5e50f81fcd73 : SUCCESS in 1h 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7048163eb68a470eaefe65c2fd1d0fb9 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9f0d8ba35c4812a7acf2621e1531d5 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8006b946b68541d39fc4e9f1dc63d4f6 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ca2d50bb212434daffcc21ecc8af39f : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24532cba48847feb32d3d26f75dfbaa : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2e569a748ed400aa3d3b2a8dbf90d83 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2145438dc0124895b1c74d6c7a6f6e1b : SUCCESS in 35m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2182ab743c3747489cfe1011e781e9bf : SUCCESS in 1h 15m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/97817cd9b37649fb98ad3bebbce31966 : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":3},{"id":"588c106ff73adb8ae495e7fc0bb068f0de51505a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:52:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56425faeaeb04b85ba0580597be2e04c : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/052dd68e9d2741ef977b3a97dbc05ead : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0cfd36292cccb8d3af9c65a1dff9696f50870df6","revisions":{"b164bd8e90e954fecec52874ec0202e92e5f55de":{"kind":"REWORK","_number":1,"created":"2020-07-06 09:32:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/739440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/739440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/739440/1"}}},"commit":{"parents":[{"commit":"acf1600973a2e6c239538384a34feb41d080141e","subject":"Merge \"Bump prometheus-openstack-exporter version to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf1600973a2e6c239538384a34feb41d080141e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:31:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:32:20.000000000","tz":120},"subject":"Remove sensu images","message":"Remove sensu images\n\nThey have been deprecated in Ussuri [1].\n\n[1]: https://review.opendev.org/#/c/711636/\n\nChange-Id: I0cfde46c29eedce77c70c1c677220814a801ff3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b164bd8e90e954fecec52874ec0202e92e5f55de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b164bd8e90e954fecec52874ec0202e92e5f55de"}]},"branch":"refs/heads/master"},"d04efad3a4b6e389e7d13e78b0c0a5ea82f3f1f5":{"kind":"REWORK","_number":2,"created":"2020-07-06 09:35:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/40/739440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/739440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/739440/2"}}},"commit":{"parents":[{"commit":"acf1600973a2e6c239538384a34feb41d080141e","subject":"Merge \"Bump prometheus-openstack-exporter version to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf1600973a2e6c239538384a34feb41d080141e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:33:24.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:35:07.000000000","tz":120},"subject":"Remove sensu images","message":"Remove sensu images\n\nThey have been deprecated in Ussuri [1].\n\n[1]: https://review.opendev.org/#/c/711636/\n\nChange-Id: I0cfde46c29eedce77c70c1c677220814a801ff3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d04efad3a4b6e389e7d13e78b0c0a5ea82f3f1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d04efad3a4b6e389e7d13e78b0c0a5ea82f3f1f5"}]},"branch":"refs/heads/master"},"0cfd36292cccb8d3af9c65a1dff9696f50870df6":{"kind":"REWORK","_number":3,"created":"2020-07-06 09:50:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/40/739440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/739440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/739440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/739440/3"}}},"commit":{"parents":[{"commit":"acf1600973a2e6c239538384a34feb41d080141e","subject":"Merge \"Bump prometheus-openstack-exporter version to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf1600973a2e6c239538384a34feb41d080141e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:33:24.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:50:12.000000000","tz":120},"subject":"Remove sensu images","message":"Remove sensu images\n\nThey have been deprecated in Ussuri [1].\n\n[1]: https://review.opendev.org/#/c/711636/\n\nChange-Id: I0cfde46c29eedce77c70c1c677220814a801ff3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cfd36292cccb8d3af9c65a1dff9696f50870df6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cfd36292cccb8d3af9c65a1dff9696f50870df6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
