)]}'
{"id":"openstack%2Fkolla~739446","triplet_id":"openstack%2Fkolla~master~I6d3174e8b52cb4d1f2eca4d85d506665dafe2406","project":"openstack/kolla","branch":"master","topic":"remove_opendaylight","hashtags":[],"change_id":"I6d3174e8b52cb4d1f2eca4d85d506665dafe2406","subject":"Remove opendaylight image","status":"MERGED","created":"2020-07-06 09:48:57.000000000","updated":"2020-07-09 17:25:40.000000000","submitted":"2020-07-09 17:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739446-1594315305837-f967fff8","meta_rev_id":"c26ae30299539362777df20130cf59d6f12dee82","_number":739446,"virtual_id_number":739446,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-09 09:51:28.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-09 17:21:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-09 13:29:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-08 10:58:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-07-09 09:56:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-07-09 14:41:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-07-09 09:56:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 08:29:25.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-07-08 10:58:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-07-09 09:56:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-09 13:29:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 14:41:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-09 17:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a3971c3f87e2fe9492ce175b06542a70f5cf01f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-06 09:48:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8171ce0b87a7902f60a1ca73e4764b66628f794","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-06 10:21:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5fa8bc7dc50b4fc99c86744d421cd1aa : SUCCESS in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"093e5ffeae5e8fd98262bd6e283877f12e1af578","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 10:29:02.000000000","message":"Patch Set 1:\n\nDuplicate of https://review.opendev.org/#/c/734630/","accounts_in_message":[],"_revision_number":1},{"id":"1b308471e3a44254e180b219ac1148c11ada49f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 10:29:43.000000000","message":"Patch Set 1:\n\nThis one has a slightly better reno...","accounts_in_message":[],"_revision_number":1},{"id":"b61448efe5fb78dfef33fb6f19f7ce57d14931f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-06 10:35:28.000000000","message":"Patch Set 1:\n\n\u003e This one has a slightly better reno...\n\nShould I rebase the earlier change and make it better, or should we merge this one? :)","accounts_in_message":[],"_revision_number":1},{"id":"dd44757290505ea3307afc526138648c8db1fa8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 10:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/adeca4af3ee540409001d72d1e0527be : FAILURE in 51m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/860fc3ff23784afabbc8d24bbc887570 : FAILURE in 42m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a8bbb7cf0c94b3785f201113483af7b : FAILURE in 59m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88bddaac79e04bdda6f95ad33f96d459 : FAILURE in 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c2d323570128ee0b0c30eea6045911f9bda82e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 11:00:30.000000000","message":"Patch Set 1:\n\n\u003e \u003e This one has a slightly better reno...\n \u003e \n \u003e Should I rebase the earlier change and make it better, or should we\n \u003e merge this one? :)\n\nI don\u0027t mind, just pick one and abandon the other :)","accounts_in_message":[],"_revision_number":1},{"id":"6a7f4c1625b89fe18b6a32e0503d4e86cfb42072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:05:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e8e164764cec4cb898b3b39b8edb46e6 : SUCCESS in 48m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/899ba407b0514d06832f03752f589ed6 : SUCCESS in 46m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/53327158da3d4a3eb86cb86c4799d441 : SUCCESS in 1h 30m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/41bd6b9e40ea461f8fda5cc720a562a0 : SUCCESS in 1h 15m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bce82cd579d043a5ab7b18072e8924d9 : SUCCESS in 1h 00m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fd50908706624fe9be0d7556d33e399a : SUCCESS in 47m 03s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/19cb6d72e6634435af938b7642dabf31 : SUCCESS in 35m 34s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9fa977d2e1d34be689e3d0014c54ba7f : SUCCESS in 1h 46m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/044588f3135244acaafed9b2bc1c2823 : SUCCESS in 1h 37m 18s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/93cfe9b9c0a24fe88f374aec21adcf64 : SUCCESS in 1h 50m 10s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/10eeda727c5a4f228d989b7d55fff0e0 : TIMED_OUT in 2h 08m 22s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/eb058514e9044aeea90a68e62e1e1ff8 : SUCCESS in 1h 24m 35s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/063a8a28e8ce45aea1fed8dc5afd5b73 : SUCCESS in 1h 53m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c98ab663e3b44615948d4d337404262c : SUCCESS in 26m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/befe5e16b29349ad9dea31f9d9928cb1 : SUCCESS in 55m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a5df013206be4b6fbab132385660406b : SUCCESS in 1h 02m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94690503df5945bdae3d89d15f057be8 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1987a915c3459dab01facb3f5ac836 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547e7b5d65a746c896695dc0c038bf91 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/207638c1471d4960a108a5bfcc7adc45 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd3b9851ae141b890c6d149a4ed3889 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9db201e75114d74b74956f1c17a1d33 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3cd7e28381d24d9c88a974edbc17913d : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d330258c9ce403f924edc39746b7890 : SUCCESS in 39m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d77d0a3fff7b46e1905b49b58064f1a7 : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e00a50379674b77a2a1e42f658fe5bd : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d5e6c859af2426385296163dbdf3822 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5c2055e070c44b70b24639d3974adff5 : SUCCESS in 1h 36m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/43d689819fd74b5fa592c9da6f3a8edd : SUCCESS in 2h 03m 56s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8191c53f11ef4788bbcdf5b7a736f4fc : TIMED_OUT in 2h 07m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4a8c74b53ea840aca9eabe9edb772789 : SUCCESS in 1h 57m 47s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8d0bf5cd7ac748559f62fa173f7bdb75 : TIMED_OUT in 2h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e941e21e174702ed951dae0cd8d0e19654f1b041","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 12:52:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"742ea65610bb8a1dcffa5174cb4a101308045610","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-07 07:36:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c6ac015f21c95664c96d66c87489d58fa3eaf96a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-07 07:36:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\npurrfect grammar","accounts_in_message":[],"_revision_number":2},{"id":"2da0c3c8c8cf33a4b6a63ffa22b4e91e24b0d4be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-07 08:03:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0bc0e12e86b4497fbf3208ba44a25c7b : SUCCESS in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bfae1432a4aed5982694049330b863f78530f7e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-07 08:06:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4ae483bc41608d52fc692228b24879c75fc0e69b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:13:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6688ad647f31479a891f30715aeb1448 : RETRY_LIMIT in 2m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d6d5e2775aa4760826b46594eaeab36 : RETRY_LIMIT in 3m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a40fcb4debbd425487b5df243d02c139 : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f72bc82f4f724340a76be91dbedebb82 : RETRY_LIMIT in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ffcea3a1c70f800be5278d9d75c43b2550b417d","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-07-07 08:29:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1b147cec16866eb9b6af8a62cb69db390162ee1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:57:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/53c1b667753e455bb03a0cb9f9661227 : RETRY_LIMIT in 3m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6fcc4e7149a4f18aba9fb78be693044 : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6cd8581ebf84171b925ce8a23a44e6a : RETRY_LIMIT in 2m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4622287127c942b4a9f65f14f6bb3c74 : RETRY_LIMIT in 2m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3967631be3ef06838d61207dbba46d8839df9223","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-07 08:59:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/dcce6f34072447e189fc09f751f17d99 : SUCCESS in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31eb7fa1ca60e125f3ff3e5efc51018e8e684adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:42:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bcdde9bc7b33426b811a0b602e2ffc58 : SUCCESS in 23m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/77383b5b8d824a6fb8bded510f79a2a3 : SUCCESS in 45m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/61edb59f3ac149ee8eb4784a8a413bf4 : SUCCESS in 1h 20m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f5da3131a4b8403ba5d9e8dea810552a : FAILURE in 47m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa29bfca39734c43b8e9abdb87e5a4b3 : SUCCESS in 1h 22m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ae3149b2c60c424798a77ee715ba2117 : SUCCESS in 34m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/83a1c76add3841d9a3db1aab8247601e : SUCCESS in 40m 44s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c1cfcb23e6b44f5db26673bd47e452b3 : SUCCESS in 2h 27m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b0f9d90469f047ef90306898c88e06f3 : SUCCESS in 1h 56m 32s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/af5577aa19db4b0c995b2559b41d8b98 : SUCCESS in 2h 00m 03s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/60267a85674841c68e86a5ebd49ad8f3 : SUCCESS in 1h 47m 35s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/5dd55edc1bbb4c4e9bca88d8b4400f82 : SUCCESS in 1h 12m 29s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/7d4ed49d8bb34ca698206bc747ae39d9 : SUCCESS in 1h 50m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/03b08a1fb4544279902b554aad5bcca1 : SUCCESS in 37m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b00a8657f2345b68c469d43630ca071 : SUCCESS in 18m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37cf236de3f8404e9f3cb117ab565f3c : SUCCESS in 1h 03m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c486621f4f844c1987c863eeb32f8e21 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55da9b39a2e4445b14d6a2d7294c25e : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece4c81199044fd88ecd780d6cf41fd1 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42441f085b7d468a8535b3f87e3138eb : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1666b07358248a3a47dd2c37c6e5afe : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/161b03b6beca49e9acc634bb958bd912 : SUCCESS in 7m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3db019c8a1ca4e889e9b87e2a9dd2f7e : SUCCESS in 23m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e33a5078fd594ed2984d14d9017d85eb : SUCCESS in 30m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfe9cd777be043c39e9aa2bb1d7e2ded : SUCCESS in 1h 06m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a5badf85a7f40a9815734ce04803443 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d71454201f0b418ea01bb5515562b46b : SUCCESS in 1h 05m 02s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9bf2e6070fae4acebfaa8c0d16944e43 : SUCCESS in 2h 10m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f4edca4d5c6f469b99c8ab06ad6e6a3c : POST_FAILURE in 1h 43m 32s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1af21c402ecb43d1b974f8ea19302cdd : SUCCESS in 1h 36m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ee65eb87df8d45dcaf9e63eb7df1ac2d : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/7368af45ad364481ae7d55b85bd9d4b5 : SUCCESS in 1h 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f3796d44acc216eee5247576b25078051c0704e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:42:14.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"864122f77c4a4e46a8c1fc11adb2b76022fead05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-07 10:51:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e80f5ff9b01eded128e63ff3f1221564accc46eb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-07 11:19:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f7753408ed9e488e831f64433a42b76d : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"096749cf036c0c7e1297ce94b3148ddcef3b1cc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:34:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef507d6493604f13bdce1fd14d71fa8d : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b46d3f2f01c94afaa51a9f1692e33ae4 : SUCCESS in 1h 28m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ec91cf80a0f43c29628f15390696ff4 : SUCCESS in 2h 38m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d076b58b66a14c178e64f84b923d4a3e : SUCCESS in 1h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b78cdc44b22cb898a7c41468a2894c73d3bac3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:36:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2e80b93467b5444fb27fd267f461d00c : SUCCESS in 22m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cf03ed8c30c94bbaa181154869caa294 : SUCCESS in 26m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/69ca8b77bbac4b02bfc946ef9a293aba : SUCCESS in 1h 11m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eb4319723dfe48319e5789c29ab517c5 : FAILURE in 42m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b69328c2576344709ea7f662045239e2 : SUCCESS in 1h 21m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9d393dadb81341eaa6043a6e85a586ff : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/c5d25c03078848c8b30760d76a705cc3 : SUCCESS in 31m 57s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/86ebcc09c8634082bef26d79aa4077f1 : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1c8562e208d8412396f0c7a141312349 : SUCCESS in 2h 42m 54s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6643871fd1634686928ad43be242758b : SUCCESS in 1h 54m 58s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/75e3d9cdd7d2445ca771664c9320da7c : SUCCESS in 1h 42m 47s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a3807ef9187c4df39928f3e169557ec1 : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/c4db5854b3814c759122962b28e7d8f8 : SUCCESS in 1h 42m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ff88f8883e9454baa09f37dd6819102 : SUCCESS in 19m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a47db08b1d8e44e2a53cdb980eb0ea24 : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7a29ff083fbe49cea2d791b8fdeda56b : SUCCESS in 1h 01m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9ce2ac546da44858ea3497108785524 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c12116ecc154a3baa70f406a7a7f913 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/235ce0fdf315416c9af4ad05be4b605c : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bda8631d608446292c6b8e9770e2000 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc5398fb1e9409ca677eaf20a6c5606 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8399bc92dfe140c3a039de8e30624051 : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7ecd25478234a0582c6b4d85fb5d0b8 : SUCCESS in 26m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e04b1935b3504a209275319e51af941a : SUCCESS in 25m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e545144340941659ae9f04fd6265024 : SUCCESS in 1h 01m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd2f5acaab5d488aa70d5eae7d125d6f : SUCCESS in 49m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e4e16bcd6ffb45408916f8b0d76f9746 : SUCCESS in 1h 11m 00s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/36738a352278402e9cfd67a2bd9e40d4 : SUCCESS in 1h 41m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f61aea3593364635ac4f83eb71b833d8 : POST_FAILURE in 1h 45m 27s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e67f495fe2de4b7f879d0d8f381a9f32 : SUCCESS in 1h 47m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/af7754dc93cc42dd82ee615773c9c870 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/fd1c64568a2947cca75a3c94a7c1e66c : SUCCESS in 1h 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8dcba2294c832650d79f97e9ea20743deb343b28","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-08 10:58:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"db7cf060d6e70efaee10192130c5cb6fc9d51531","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-08 11:21:15.000000000","message":"Patch Set 3:\n\nAre we sure that Tripleo aren\u0027t using the ODL image?","accounts_in_message":[],"_revision_number":3},{"id":"b40c4789a8a348deafc89a7191efbbea8c908aea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-08 12:08:17.000000000","message":"Patch Set 3:\n\n\u003e Are we sure that Tripleo aren\u0027t using the ODL image?\n\nhttps://github.com/openstack/tripleo-heat-templates/blob/master/releasenotes/notes/deprecate-opendaylight-service-64b960923324edc4.yaml","accounts_in_message":[],"_revision_number":3},{"id":"63c7fc866cde7fd63c9faa4a518ef2b535860549","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:36:24.000000000","message":"Patch Set 3: Code-Review-1\n\n\u003e \u003e Are we sure that Tripleo aren\u0027t using the ODL image?\n \u003e \n \u003e https://github.com/openstack/tripleo-heat-templates/blob/master/releasenotes/notes/deprecate-opendaylight-service-64b960923324edc4.yaml\n\nDeprecated !\u003d removed","accounts_in_message":[],"_revision_number":3},{"id":"38da7174a83f329802f455c027557065c6eb100d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-09 09:51:28.000000000","message":"Patch Set 3:\n\n\u003e \u003e \u003e Are we sure that Tripleo aren\u0027t using the ODL image?\n \u003e \u003e\n \u003e \u003e https://github.com/openstack/tripleo-heat-templates/blob/master/releasenotes/notes/deprecate-opendaylight-service-64b960923324edc4.yaml\n \u003e \n \u003e Deprecated !\u003d removed\n\nhttps://github.com/openstack/tripleo-heat-templates/commit/9c3638b8a13ea08a1bab95c9cd6ab335b729b6b3\n\nHappier? :)","accounts_in_message":[],"_revision_number":3},{"id":"081dbe101d41752bc51a7985fdc8f20a0d6a96b7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:56:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n\u003e \u003e \u003e \u003e Are we sure that Tripleo aren\u0027t using the ODL image?\n \u003e \u003e \u003e\n \u003e \u003e \u003e https://github.com/openstack/tripleo-heat-templates/blob/master/releasenotes/notes/deprecate-opendaylight-service-64b960923324edc4.yaml\n \u003e \u003e\n \u003e \u003e Deprecated !\u003d removed\n \u003e \n \u003e https://github.com/openstack/tripleo-heat-templates/commit/9c3638b8a13ea08a1bab95c9cd6ab335b729b6b3\n \u003e \n \u003e Happier? :)\n\nYes!","accounts_in_message":[],"_revision_number":3},{"id":"f435937cd7454a97f265ffcc8d47cdd3be99ad12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:56:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7873892d441b0dc9f13cd1226361f09e860148ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 11:16:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d3519a60b2d14fe3982c4348e0f17a73 : SUCCESS in 36m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/452970eb4d51410c980230cbdccb0e29 : SUCCESS in 1h 00m 56s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/141ef2d451d04d3999f78ecda6b8599b : SUCCESS in 1h 17m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/498a31dadef94c62b8e36fa5656fd4bc : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f689093b3b4dc0b70d44fa68ac4474 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5716c3c58846d8b79d25f41cbd78d7 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d8f911f3f814b44828413738c2e6423 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ebec35c63d4ea2b1a09aa563093f1c : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41abbe902fd64f4fa410a01a2dad3b87 : SUCCESS in 7m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/351b737d8430497293bfdab4adb5f057 : FAILURE in 31m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6494a238927a445aab6eb38654d03c5b : SUCCESS in 1h 00m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/207f4848f44c4ccfbafcf45293b2fad3 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ea535ae4cf1918637f70775204cda1e4c2ae1aa6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 12:58:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9007877c4161e15a9b20c6667165679b5e4f5f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 13:19:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c2469b0248d4a94be03a2ef23d8f3af : FAILURE in 7m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9daa4c9615fc415790d7f7cb27e75d03 : FAILURE in 7m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a2eef4f1bafc41149b8736dd6a397fdb : FAILURE in 20m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dd04f1163f7f4f00bb79820a14133b95 : FAILURE in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf058efa38b8ba61ed5724f020e9784b2427b1d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-09 13:29:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/2436f054c54b4c86bf903240de0b7209 : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cf94181aafe04732123fe80188a74af0c0e56b1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-09 14:41:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"42d821c95d5fc98d526af020fb0b01afe55b855b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:36:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2c540ca8046141c7b9588547db59c1ac : SUCCESS in 46m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a8a0e77352724894858cf9d0f76ddfc4 : SUCCESS in 49m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c94252cd541a4ead916df02bd103ef86 : SUCCESS in 1h 17m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1fcc27e2adf946e493d571384c755191 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/56ad7aac8cdb4a998064c1b59e4a2ae3 : SUCCESS in 1h 23m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bf7135bb7c9447958d688f668a00c580 : SUCCESS in 39m 26s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d0d76de09bea43a99942a00e35c85f5b : SUCCESS in 31m 18s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ad49bb93e996485bbee39fab7986b16a : SUCCESS in 1h 52m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/364dc0c6fd764127a6e666a60e65908d : SUCCESS in 2h 32m 58s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f4c4c69ad232464a97e4a9ec25b1a444 : SUCCESS in 1h 58m 44s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9941573a18c241d19d78a1dd27719292 : SUCCESS in 1h 55m 27s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f6dfd81ddd7a48d4b29be84ee4eb4630 : FAILURE in 54m 25s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/9bd2c3ce9f1c49448341805eb1e1ff2e : SUCCESS in 1h 56m 01s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b4c1f90f3874d81b5c8207444355578 : SUCCESS in 38m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/81ec4f69e50f41faa9612448c735e344 : SUCCESS in 26m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b7c452ed3444e92a9fa4695331e0e69 : SUCCESS in 1h 03m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9aef6096e9f4415dace854ae09090e55 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27adce050d92492fb25ba07c04c12bd8 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838edd11a3c14c1a9ca6de866ac5f226 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b417be04fbdf42d28fb37689c1aeb058 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66d43edec2042529fb1bfcf257cddd2 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cf0a9cb23434e69af9dc8520646d5fe : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9e3f29a51034138affb8f853610d2be : SUCCESS in 35m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38a62e0246674bc5813a5c8b68c0cc60 : SUCCESS in 25m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/517d16ee8b3349f8ab24b71b7be5a450 : SUCCESS in 1h 03m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbaf0151e5de417c8f9de9e41bd587c9 : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fcf540c0e58145f79fc028db2760a21b : SUCCESS in 1h 21m 12s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/751d9e62b67844769ee31aa7f085f165 : SUCCESS in 1h 35m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/baa9c2e338ea47a48991fc0f9eec34a5 : SUCCESS in 2h 07m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ad3da5ecd98c4f78a4224d5ecd2fc5e8 : SUCCESS in 1h 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/37b71d554228403d9a3e51bd38d94f84 : SUCCESS in 1h 08m 04s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/90d630fac8f747a0a634584f7753d99d : SUCCESS in 1h 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8fa5435637e4c96f3a12c32c05f6fb41a23d764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:37:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea13f397ab720a0b92209cebaecb89b546de16dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 17:21:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/723d7e2aafa54ff3805a67377df18350 : SUCCESS in 25m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7dbe9ac70815424e97fe47338efe41c4 : SUCCESS in 1h 17m 46s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7ddfd14b8a4141c89840369586ab5c1b : SUCCESS in 1h 24m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64117cbd05964f27a1d186856aa0f3bd : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c142970dd24beb88f87b35f9231e67 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956fbf4fb0404873b61e23667a4df48e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a63a9a71a12743888b9a3e2e395d17bb : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f967e3b8410e47cdb3d57afe33e1f736 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d57cab780004aa1a979519f6f79fbff : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc0d1a9a1a9c445283df2b1545b4c40b : SUCCESS in 46m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d14d5a9e5f4841e299e5cabb5e8da76b : SUCCESS in 1h 00m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/addfb469566f4eefadccd8e19246a10f : SUCCESS in 1h 40m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8109da9eda9de7fd4a57acb100b07ff497630959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 17:21:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"eb4bb118392c3e3601535a5acd16b36e4e18c170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 17:25:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4733c8c4ebad482cbd2b2f3bb5ab34aa : SUCCESS in 2m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eecd0a8bb8e44259b65d66aa08f0b6e2 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6c77d52cda63cc51f6e15943b1733a5100b08392","revisions":{"9bc8b8a4e1a3b0c0e875033fa14794d381c3c175":{"kind":"REWORK","_number":1,"created":"2020-07-06 09:48:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/739446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/739446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/739446/1"}}},"commit":{"parents":[{"commit":"acf1600973a2e6c239538384a34feb41d080141e","subject":"Merge \"Bump prometheus-openstack-exporter version to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf1600973a2e6c239538384a34feb41d080141e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:48:24.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:48:53.000000000","tz":120},"subject":"Remove opendaylight image","message":"Remove opendaylight image\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I6d3174e8b52cb4d1f2eca4d85d506665dafe2406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8b8a4e1a3b0c0e875033fa14794d381c3c175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8b8a4e1a3b0c0e875033fa14794d381c3c175"}]},"branch":"refs/heads/master"},"55ac130200b18cab0af59bcef0242980dc20462c":{"kind":"REWORK","_number":2,"created":"2020-07-07 07:36:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/739446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/739446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/739446/2"}}},"commit":{"parents":[{"commit":"acf1600973a2e6c239538384a34feb41d080141e","subject":"Merge \"Bump prometheus-openstack-exporter version to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf1600973a2e6c239538384a34feb41d080141e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:48:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-07 07:36:08.000000000","tz":0},"subject":"Remove opendaylight image","message":"Remove opendaylight image\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I6d3174e8b52cb4d1f2eca4d85d506665dafe2406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ac130200b18cab0af59bcef0242980dc20462c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ac130200b18cab0af59bcef0242980dc20462c"}]},"branch":"refs/heads/master"},"6c77d52cda63cc51f6e15943b1733a5100b08392":{"kind":"REWORK","_number":3,"created":"2020-07-07 10:51:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/739446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/739446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/739446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/739446/3"}}},"commit":{"parents":[{"commit":"4f4b0126a9316b2746306eec86c1224f220aef1c","subject":"Merge \"Remove sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f4b0126a9316b2746306eec86c1224f220aef1c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-06 09:48:24.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-07 10:50:38.000000000","tz":120},"subject":"Remove opendaylight image","message":"Remove opendaylight image\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I6d3174e8b52cb4d1f2eca4d85d506665dafe2406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c77d52cda63cc51f6e15943b1733a5100b08392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c77d52cda63cc51f6e15943b1733a5100b08392"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
