)]}'
{"id":"openstack%2Fkolla~739507","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I81e5b7abf4571fece13a029e25911e9e4dece673","project":"openstack/kolla","branch":"stable/ussuri","topic":"disable-systemd-getent-stable/ussuri","hashtags":[],"change_id":"I81e5b7abf4571fece13a029e25911e9e4dece673","subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","status":"MERGED","created":"2020-07-06 14:44:54.000000000","updated":"2020-07-08 09:30:14.000000000","submitted":"2020-07-08 09:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739507-1594200395118-5d897941","meta_rev_id":"e4cd3d97d5bf27d87023fecb85d5b57811d0aa87","_number":739507,"virtual_id_number":739507,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 09:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-07-08 07:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-06 14:45:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-07-08 07:45:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-07-06 14:44:54.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 14:44:54.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2020-07-08 07:45:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-08 09:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae289e114b9625fadc45e93b12d30b9fbd563c11","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:44:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dcad18327186d40405e4f140f6ccc2c0f8e934ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 0e6465543e4ae91c9b791944444e72905a716a6d","accounts_in_message":[],"_revision_number":1},{"id":"cc010296c1dcb72f44d869117a76461a2712ac6e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27aa62e02abefbc9905cc4600d3be98f1dfa2c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c97f3e721e3437283d02035b2b20466 : RETRY_LIMIT in 3m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0c8c1f4c3dc4ec8861dc0c5b3cbb2f0 : RETRY_LIMIT in 2m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2564f94be80a47a391b60911438ff663 : RETRY_LIMIT in 2m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/09f8036ddc34476ea8ccde724596ca53 : RETRY_LIMIT in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92e8b8783e90379308b9d89b24828242681080d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/89821ceb84a741cb9e00894f7b91de56 : SUCCESS in 29m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce4d57f7919a4109b42d2c62b6d98774 : SUCCESS in 31m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b4e09ccd1b8d4341bd1bb349d7538a99 : SUCCESS in 1h 24m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a00f501a785d4ed9bef7f90132118898 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/347be56e7c8e420cbce77e9126996606 : SUCCESS in 1h 33m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/176fee35f30048c2adc9d8897a7eb6e9 : SUCCESS in 58m 25s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3e4232008d5e4addb7b34f0af8a29c32 : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7eb755df6a504c358678ac30cfaebc85 : SUCCESS in 2h 12m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e4f877f654de43509816425238715088 : POST_FAILURE in 1h 08m 46s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b85dce9c121441ebaf42b3434f497a37 : TIMED_OUT in 2h 12m 23s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/caba02c820b54906903cbb1094450b7d : SUCCESS in 2h 02m 45s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/00ce77d4e9754b15ba9f9ca17b0dc915 : SUCCESS in 55m 49s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/c63611d88feb4195a0c4a4f0eef67b92 : SUCCESS in 1h 57m 03s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/05c23113ad0e42758a823fd922e4e457 : SUCCESS in 22m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c30f2880b6b941fc96a879ed1a98def3 : SUCCESS in 41m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4848b53b16c49e3a315003ad9b27a74 : SUCCESS in 1h 31m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58194d8b726946cc9c779ec463d33985 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875439946ca3419081a162575e26f280 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e292a9a7ab24f3fa369977e7edf830d : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dcf7a00e3084f9b9b3911cb35860d06 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f071327d634dd696a8c52fb21625e5 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3802a4f307447d6ae9c41696a457280 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25a6bf6e0d1441c289640c45db677476 : SUCCESS in 29m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd119a497b97436c9d3029f74530c377 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac821b08c31f4ecabee750ebf175e2ae : SUCCESS in 1h 26m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00d2f1cca3a34043b12d979e8e7a6474 : SUCCESS in 1h 10m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ff26c81ed804cc58d5d1c7caedee7e8 : SUCCESS in 1h 35m 38s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/03d6a037c02142fb9605ada2203fe0da : SUCCESS in 1h 39m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e39ce45d4d93440294d65c180aff7c9b : POST_FAILURE in 1h 44m 26s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0b12ed0d99634453898b9121056ba10f : SUCCESS in 1h 56m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b1f4249b45454e34ac0b122798656d22 : SUCCESS in 1h 11m 42s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8aa71f3af4664101a4e3f99a73a510fe : SUCCESS in 1h 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"160c3f89d88a35044dddb7cc3b29aaa8b0a79866","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-08 07:45:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0811a5414f0d8a3c05b04fae1e1aa4c373265d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 07:46:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"279902c1c257420024476f5ac466a56587db0b14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fc4da421b07a4c2092671db0b7c3dae3 : SUCCESS in 31m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a7b208ef5da54f2c80006e5d564ad138 : SUCCESS in 1h 16m 59s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/17209db2b32e42f28668dfabb48fca5a : SUCCESS in 1h 24m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7f15e23dfd848dabb4ce83e85c74e76 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835590e8c79b40f4adaeaad7576f9e9d : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90338a596824440b97804b40ec300818 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e43aafbc3674ea6aa3fa12658784538 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f524b912dca446578e14d2df5056088f : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f6d4d4a66b34ffba8a9f24e33e9607b : SUCCESS in 6m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d7890260fc543218b35c8a946ba608f : SUCCESS in 29m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31e5f7a13a6044008a276e345b4a72d6 : SUCCESS in 1h 08m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8f368abb92f241c5a18eb0dbb881bb52 : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3e2ba7ff4f38c88105e09807c1755d5fe0d5d944","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b650231d19858d2efd6b7317f86369491d1e109c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:30:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3874edfc14240e9b41570d0d1c3d6d6 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce8f75995f724eddb79f11fa7be1c3a1 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e667c10c67f7bbbc7682c775a37b35ad064697ad","revisions":{"e667c10c67f7bbbc7682c775a37b35ad064697ad":{"kind":"REWORK","_number":1,"created":"2020-07-06 14:44:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/07/739507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/739507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/739507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/739507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/739507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/739507/1"}}},"commit":{"parents":[{"commit":"be5da6d2c9b595cfc49e2c744d1464120f24e99e","subject":"Bump prometheus-openstack-exporter version to 1.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be5da6d2c9b595cfc49e2c744d1464120f24e99e"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-06-11 09:47:02.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-06 14:44:54.000000000","tz":0},"subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","message":"Drop systemd support from nsswitch.conf on RHEL-based distros\n\nA bit like we did for I3e0e86026f5a4a78473bed824cd1682d3a020cd5 we\nshould remove the nss-systemd lookup from containers. The reasons for\nthis are as follows:\n1) Just like for I3e0e86026f5a4a78473bed824cd1682d3a020cd5\nwhen this nss module is triggered it tries to talk to dbus.\nIt triggers a bunch of selinux denials and it makes little sense\nto open all containers to talk to dbus.\nIn particular, if a container is run as non-privileged and bind-mounts\n/run from the host, we will hit selinux denials like the following:\n\n  type\u003dUSER_AVC msg\u003daudit(1592337775.860:74119): pid\u003d1284 uid\u003d81 auid\u003d4294967295 ses\u003d4294967295 subj\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 msg\u003d\u0027avc:  denied  { send_msg } for msgtype\u003dmethod_call interface\u003dorg.freedesktop.DBus member\u003dHello dest\u003dorg.freedesktop.DBus spid\u003d406228 scontext\u003dsystem_u:system_r:container_t:s0:c162,c886 tcontext\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 tclass\u003ddbus permissive\u003d0  exe\u003d\"/usr/bin/dbus-daemon\" sauid\u003d81 hostname\u003d? addr\u003d? terminal\u003d?\u0027UID\u003d\"dbus\" AUID\u003d\"unset\" SAUID\u003d\"dbus\"\n\n2) It just makes little sense in a kolla-world to have containers\ntalk to dbus/systemd and it saves us some time when a lookup triggers\nthe systemd module for whatever reason. Especially because the\nnss-systemd module does a few things which are not useful in a container\n(ensures that the root and nobody users and groups remain resolvable,\nSystemD\u0027s DynamicUser\u003d feature, provide Lookup API via Varlink)\n\nThe sed regex gives us the wanted results:\n$ diff -u /etc/nsswitch.conf.orig /etc/nsswitch.conf\n--- /etc/nsswitch.conf.orig     2020-06-19 07:18:10.974580755 +0000\n+++ /etc/nsswitch.conf  2020-06-19 07:20:12.260230103 +0000\n@@ -53,9 +53,9 @@\n # group:     db files\n\n # In order of likelihood of use to accelerate lookup.\n-passwd:      sss files systemd\n+passwd:      sss files\n shadow:     files sss\n-group:       sss files systemd\n+group:       sss files\n hosts:      files dns myhostname\n services:   files sss\n netgroup:   sss\n\nRelated-Bug: #1883849\n\nChange-Id: I81e5b7abf4571fece13a029e25911e9e4dece673\n(cherry picked from commit dc2ddfa9750f94430a5b04b62fc1fa6adcb234f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e667c10c67f7bbbc7682c775a37b35ad064697ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e667c10c67f7bbbc7682c775a37b35ad064697ad"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
