)]}'
{"id":"openstack%2Fkolla~739508","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I81e5b7abf4571fece13a029e25911e9e4dece673","project":"openstack/kolla","branch":"stable/train","topic":"disable-systemd-getent-stable/ussuri-stable/train","hashtags":[],"change_id":"I81e5b7abf4571fece13a029e25911e9e4dece673","subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","status":"MERGED","created":"2020-07-06 14:45:03.000000000","updated":"2020-07-08 09:30:03.000000000","submitted":"2020-07-08 09:26:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739508-1594200393216-86bf99d6","meta_rev_id":"8e1d358c4b023c80a734207c9465348495e30122","_number":739508,"virtual_id_number":739508,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 09:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-07-08 07:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-06 14:45:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-07-08 07:45:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-07-06 14:45:03.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 14:45:03.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2020-07-08 07:45:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-08 09:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c055a50351f1dd8a74b5bd4d8c8afefe6c76fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"bb3ccb205213f799fbd5a27ae046b0b63836d4b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:10.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7038e3ff03c6d08e7b794f9087675964ddbbed0a","accounts_in_message":[],"_revision_number":1},{"id":"b06166b4223cd854d22c56d80fe348c85644484e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0fcf5f707e16ddecf48b296884a83b0d7e02fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:29:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/527c78ed382f42e087f3e4e58785a00a : SUCCESS in 48m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6f0a31df8e174556bdb6709e163683b1 : SUCCESS in 51m 12s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3e03dcd9d0f3449da9d5ce5c565add1a : SUCCESS in 1h 17m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bfadad93136d4c9a98c0763874f482fb : SUCCESS in 57m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9ca8431aedfb4d3fa2c79faa072b000d : SUCCESS in 1h 44m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2d1778eeb1a540ecb2f7d8435aefa4f2 : SUCCESS in 1h 28m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/89a87b38d0be46349aea2e9d86aede3d : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c7525464dab545829853ac0e699dff91 : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/08aefbf42f154279bec0c989485cbef4 : SUCCESS in 1h 28m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1695e89ca11345818bc2f0d9946cfa19 : SUCCESS in 58m 00s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/6d48f112678b4fd1b10f2285ff9dc9f6 : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/26ff9ffdfb2f42b29965531a3cd56fef : SUCCESS in 36m 23s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5c81201c00be471088ae500cf6fffc85 : FAILURE in 1h 56m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/72bf8b5efba145cfbf4d135438e76835 : FAILURE in 1h 37m 21s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/e88759bbf2d34dfabeb1142bbb84ccc0 : SUCCESS in 2h 02m 41s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/1d5c3b291a2b4b459c6d1b39a3967cef : FAILURE in 1h 11m 09s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/bdfbd1f7635e47a89cdcb6e2b155298b : TIMED_OUT in 2h 05m 39s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/199d46b4afe14a60994106904b5c7225 : FAILURE in 1h 15m 29s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d4a002beb41f4de88173a69332d17e63 : SUCCESS in 1h 22m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ef1c9043f0254beca88c0fd0f39f0e23 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/051329579569416b953c19916ba002ca : SUCCESS in 24m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2723cf895e504fd8abf5577fcdb47f35 : SUCCESS in 25m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0756f98bc997472a8763c947006227f8 : SUCCESS in 1h 03m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/318441a21ec840ea8de4b323e653c7f6 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d749f989b2641e19971171b28b78355 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96951cfa48244ebbb029b42bf2c99247 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309ed83703af41d2baf5f41b5523f76c : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9000e9c5380c4ed0b44004f2d4b67b25 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa05e500ed0c4c91b117c452b396ea6d : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf308cf03be04034913531dc64077f29 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ab8ab49571d4108aaba47cb7fd4a138 : SUCCESS in 34m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c503e5fe50d84e999f9bb842da7dad02 : SUCCESS in 1h 37m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c5ce4212ddd4f4bb101ade7614ce702 : SUCCESS in 58m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99be5e3ee35b48ffb353a95e2cdf248c : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52fab66a8e904537923444f59eecedb6 : SUCCESS in 1h 12m 02s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d0b739978764460c9cf0101b2ac18099 : TIMED_OUT in 2h 33m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/38d9ba36f14d463db0b0b68f1df60ce7 : TIMED_OUT in 2h 34m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d9b94712205d43c499908aef108d4f5c : TIMED_OUT in 2h 06m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/28d29b08f0a7465698a67e522db521be : SUCCESS in 1h 12m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"3fa51b213588b38afb4fe8169dd024450758abbb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-08 07:45:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4757c8f6dec09578b6c23dcaa41bb2c296c75e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 07:46:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"923d11af84b7416cd900d1f4ebac3991ba436835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aadffdd082394b3fb3f5ee4a2e922513 : SUCCESS in 56m 48s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ef80c40703754da4aab52c58a21706fd : SUCCESS in 26m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/58799110f4834f2e86d3defa5666d3a7 : SUCCESS in 1h 33m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8869d457e15f48e6b4dd06b44bdb4341 : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/276b9eb413c94101accef712dc138610 : SUCCESS in 1h 38m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3ceb4f270dbb48cb93004936f90b66e5 : SUCCESS in 51m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f82755eb715437cbce78e1779c88140 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6397e1c11756461fb635eba329a42992 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0aede0588124426d8b732582ece54be0 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea9c1f5ed114f45a4f4f3f463c2abba : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d4aba1bdb4d4b6cac0500a6149eb146 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc64030331234ef49e6fb83e54da5142 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e22ad811f2d843209718659f98175b86 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd30333e30b7410ab962aa871c497a43 : SUCCESS in 46m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff433a53e48540a4bbcad664462c19da : SUCCESS in 1h 05m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95a6f7a6b1f840aeb96eb1356752e168 : SUCCESS in 1h 11m 37s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"bdcb7e674eaae3c87f4d74d154b2064d7f5f1018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e23285ac27eaa5e867050b66cbc7960c34062bef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:30:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89006ebb24ba4622ac94c41ca272045c : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67327cbc0c9f4504ab90e592608d4572 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e6465543e4ae91c9b791944444e72905a716a6d","revisions":{"0e6465543e4ae91c9b791944444e72905a716a6d":{"kind":"REWORK","_number":1,"created":"2020-07-06 14:45:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/08/739508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/739508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/739508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/739508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/739508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/739508/1"}}},"commit":{"parents":[{"commit":"58337efa45d4279a84ba75e515f0e0deb62a3e1f","subject":"Merge \"Add Elasticsearch Curator Docker image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58337efa45d4279a84ba75e515f0e0deb62a3e1f"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-06-11 09:47:02.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-06 14:45:03.000000000","tz":0},"subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","message":"Drop systemd support from nsswitch.conf on RHEL-based distros\n\nA bit like we did for I3e0e86026f5a4a78473bed824cd1682d3a020cd5 we\nshould remove the nss-systemd lookup from containers. The reasons for\nthis are as follows:\n1) Just like for I3e0e86026f5a4a78473bed824cd1682d3a020cd5\nwhen this nss module is triggered it tries to talk to dbus.\nIt triggers a bunch of selinux denials and it makes little sense\nto open all containers to talk to dbus.\nIn particular, if a container is run as non-privileged and bind-mounts\n/run from the host, we will hit selinux denials like the following:\n\n  type\u003dUSER_AVC msg\u003daudit(1592337775.860:74119): pid\u003d1284 uid\u003d81 auid\u003d4294967295 ses\u003d4294967295 subj\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 msg\u003d\u0027avc:  denied  { send_msg } for msgtype\u003dmethod_call interface\u003dorg.freedesktop.DBus member\u003dHello dest\u003dorg.freedesktop.DBus spid\u003d406228 scontext\u003dsystem_u:system_r:container_t:s0:c162,c886 tcontext\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 tclass\u003ddbus permissive\u003d0  exe\u003d\"/usr/bin/dbus-daemon\" sauid\u003d81 hostname\u003d? addr\u003d? terminal\u003d?\u0027UID\u003d\"dbus\" AUID\u003d\"unset\" SAUID\u003d\"dbus\"\n\n2) It just makes little sense in a kolla-world to have containers\ntalk to dbus/systemd and it saves us some time when a lookup triggers\nthe systemd module for whatever reason. Especially because the\nnss-systemd module does a few things which are not useful in a container\n(ensures that the root and nobody users and groups remain resolvable,\nSystemD\u0027s DynamicUser\u003d feature, provide Lookup API via Varlink)\n\nThe sed regex gives us the wanted results:\n$ diff -u /etc/nsswitch.conf.orig /etc/nsswitch.conf\n--- /etc/nsswitch.conf.orig     2020-06-19 07:18:10.974580755 +0000\n+++ /etc/nsswitch.conf  2020-06-19 07:20:12.260230103 +0000\n@@ -53,9 +53,9 @@\n # group:     db files\n\n # In order of likelihood of use to accelerate lookup.\n-passwd:      sss files systemd\n+passwd:      sss files\n shadow:     files sss\n-group:       sss files systemd\n+group:       sss files\n hosts:      files dns myhostname\n services:   files sss\n netgroup:   sss\n\nRelated-Bug: #1883849\n\nChange-Id: I81e5b7abf4571fece13a029e25911e9e4dece673\n(cherry picked from commit dc2ddfa9750f94430a5b04b62fc1fa6adcb234f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e6465543e4ae91c9b791944444e72905a716a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e6465543e4ae91c9b791944444e72905a716a6d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
