)]}'
{"id":"openstack%2Fkolla~739509","triplet_id":"openstack%2Fkolla~stable%2Fstein~I81e5b7abf4571fece13a029e25911e9e4dece673","project":"openstack/kolla","branch":"stable/stein","topic":"disable-systemd-getent-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"I81e5b7abf4571fece13a029e25911e9e4dece673","subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","status":"MERGED","created":"2020-07-06 14:45:10.000000000","updated":"2020-07-08 09:30:00.000000000","submitted":"2020-07-08 09:26:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739509-1594200397336-9ab5830f","meta_rev_id":"50eddfe70cbb2876048590fbbe2f8b3eeb32a399","_number":739509,"virtual_id_number":739509,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 09:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-07-08 07:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-06 14:45:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-07-08 07:46:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-07-06 14:45:10.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 14:45:10.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2020-07-08 07:46:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-08 09:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c540dd0fa1abd5d33bd37a14d6b227b84e4cdd8a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"4a2111f484a4c6c25ad5e188656f3f09e202ea9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-06 14:45:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f3c956281f44b49c9c23f61dd8db1d38641e4ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:48:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b28c86438bf349dbaa9aca5910b79c3e : SUCCESS in 52m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e46f7631a0a74760a19eae5f0f7214e9 : SUCCESS in 1h 01m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/92f6894bf73e4455a45313cf018211d3 : SUCCESS in 1h 31m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6d80e669d77c4f3e8385b254ad77d454 : SUCCESS in 1h 43m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3beb845f53a64f5bb7bfc0e924238187 : SUCCESS in 1h 46m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fec583444aeb48aa85142724e12d6d4d : SUCCESS in 1h 04m 40s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7ae464987d4c473396ff7698b6e63570 : SUCCESS in 1h 01m 18s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/9d9e091ce1b04478962ed7622d39860a : SUCCESS in 1h 08m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fccf7fc5aa44ad0a88bbfc613103ed8 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449410b8b1ab48f785ea7b2de04771bd : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3fda5356d684c9fb38d49aefa220df3 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32e89d0a5e504b42a432202e814512b0 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2842f8e7297e49bfb2ff5f198075087d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1b7a8fb1b8402496c34003456e4c4e : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4203dbcfc4b14eeab76bf3607c582344 : SUCCESS in 5m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ce97044fcc54f7dbf15a7b5b28a8926 : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4d47841a37d4252ad2436a0091ce327 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e3e6ed37e734b52852d93b9f255f9a5 : SUCCESS in 1h 02m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c514cd22f92449a586128be119dfbf52 : FAILURE in 1h 27m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/626adac7ca354ba8bec7dd1cd3b192c0 : SUCCESS in 56m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"658802673aaf2cbd08f6e5a4cdc6d83349faf2ba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-08 07:46:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8522ef3d61544e896909cb00923bc296ad9fb55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 07:46:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5b73629f29f133c7578a42eac56dea6d63206a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0d5b6c6f3a344c77bfae9715789b3886 : SUCCESS in 52m 26s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7cc0576008cd4b6099f7b4adca3dfbc6 : SUCCESS in 51m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/126a5df24c6a43a8a55b3d239c6b6854 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b05756f1fbce48a5bc55d5ef1e348d99 : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15e0a348a42b41dfb82b6e55c4daf134 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958577de16aa455a98e5427bbb6b73c2 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac6a2a0347da473da930429b3fc6a6c7 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afd52dd74e44ceebc1171e63fca4962 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b2c36c82ee491c96872450b0d808ee : SUCCESS in 6m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e25cfe297fbd4857863aaf9481d1725f : SUCCESS in 29m 26s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"6a8196c05e6f0431a4c2bc6ff3c407fd4a435dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:26:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eb81187ee0307be9115ec459a6907bdc713c3d2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:30:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b64e96db11450a9a4ca48f7f3aa0d0 : SUCCESS in 1m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3dad9e77956450797f07a76d896d2db : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7038e3ff03c6d08e7b794f9087675964ddbbed0a","revisions":{"7038e3ff03c6d08e7b794f9087675964ddbbed0a":{"kind":"REWORK","_number":1,"created":"2020-07-06 14:45:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/09/739509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/739509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/739509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/739509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/739509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/739509/1"}}},"commit":{"parents":[{"commit":"c42a756920dc813aaa9a13040311335f6a7975c8","subject":"Merge \"Bump versions for Stein\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c42a756920dc813aaa9a13040311335f6a7975c8"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-06-11 09:47:02.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-06 14:45:10.000000000","tz":0},"subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","message":"Drop systemd support from nsswitch.conf on RHEL-based distros\n\nA bit like we did for I3e0e86026f5a4a78473bed824cd1682d3a020cd5 we\nshould remove the nss-systemd lookup from containers. The reasons for\nthis are as follows:\n1) Just like for I3e0e86026f5a4a78473bed824cd1682d3a020cd5\nwhen this nss module is triggered it tries to talk to dbus.\nIt triggers a bunch of selinux denials and it makes little sense\nto open all containers to talk to dbus.\nIn particular, if a container is run as non-privileged and bind-mounts\n/run from the host, we will hit selinux denials like the following:\n\n  type\u003dUSER_AVC msg\u003daudit(1592337775.860:74119): pid\u003d1284 uid\u003d81 auid\u003d4294967295 ses\u003d4294967295 subj\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 msg\u003d\u0027avc:  denied  { send_msg } for msgtype\u003dmethod_call interface\u003dorg.freedesktop.DBus member\u003dHello dest\u003dorg.freedesktop.DBus spid\u003d406228 scontext\u003dsystem_u:system_r:container_t:s0:c162,c886 tcontext\u003dsystem_u:system_r:system_dbusd_t:s0-s0:c0.c1023 tclass\u003ddbus permissive\u003d0  exe\u003d\"/usr/bin/dbus-daemon\" sauid\u003d81 hostname\u003d? addr\u003d? terminal\u003d?\u0027UID\u003d\"dbus\" AUID\u003d\"unset\" SAUID\u003d\"dbus\"\n\n2) It just makes little sense in a kolla-world to have containers\ntalk to dbus/systemd and it saves us some time when a lookup triggers\nthe systemd module for whatever reason. Especially because the\nnss-systemd module does a few things which are not useful in a container\n(ensures that the root and nobody users and groups remain resolvable,\nSystemD\u0027s DynamicUser\u003d feature, provide Lookup API via Varlink)\n\nThe sed regex gives us the wanted results:\n$ diff -u /etc/nsswitch.conf.orig /etc/nsswitch.conf\n--- /etc/nsswitch.conf.orig     2020-06-19 07:18:10.974580755 +0000\n+++ /etc/nsswitch.conf  2020-06-19 07:20:12.260230103 +0000\n@@ -53,9 +53,9 @@\n # group:     db files\n\n # In order of likelihood of use to accelerate lookup.\n-passwd:      sss files systemd\n+passwd:      sss files\n shadow:     files sss\n-group:       sss files systemd\n+group:       sss files\n hosts:      files dns myhostname\n services:   files sss\n netgroup:   sss\n\nRelated-Bug: #1883849\n\nChange-Id: I81e5b7abf4571fece13a029e25911e9e4dece673\n(cherry picked from commit dc2ddfa9750f94430a5b04b62fc1fa6adcb234f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7038e3ff03c6d08e7b794f9087675964ddbbed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7038e3ff03c6d08e7b794f9087675964ddbbed0a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
