)]}'
{"id":"openstack%2Fkolla~742776","triplet_id":"openstack%2Fkolla~master~I02b58ac571cd4eb3b9d4a814fe0cb907a3564e23","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I02b58ac571cd4eb3b9d4a814fe0cb907a3564e23","subject":"Enable mod_wsgi in Ironic API and Inspector container","status":"MERGED","created":"2020-07-24 01:55:46.000000000","updated":"2020-09-14 10:01:11.000000000","submitted":"2020-09-08 08:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742776-1599555048643-c720d834","meta_rev_id":"d780694d65d996aba01d410857ca3217dbf21f27","_number":742776,"virtual_id_number":742776,"owner":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 08:50:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2020-08-24 15:20:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2020-08-24 08:23:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-08-24 07:29:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-08 07:20:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 02:22:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 08:23:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-24 15:20:04.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-09-08 07:20:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-08 08:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae22ccc1eb9f1e1dac156f61c7c6bc0850e5400d","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-07-24 01:55:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e3b0f704bf9a9a25c4bdc4f8ddabc0888312a8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-24 02:22:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c1e100cb039443338826d814c0b138c5 : SUCCESS in 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41c1f2f32323a0b9a30989993bcd68dd96f412e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c08acec78d144ee58f2b8ef903ddc016 : SUCCESS in 53m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b36bb8d1a3847afa38851c8a34ccf8a : SUCCESS in 36m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1b0e97d0c5e40389cf715d4e8b3def9 : SUCCESS in 1h 30m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f10f1e93ff2d48ed9c70e36d69fa76a9 : SUCCESS in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26c1630cea4d305af086fc9b5df276970562836e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:47:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/edde3273d7c44d7db02a051923053f8e : SUCCESS in 21m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/54cf2f3a012840b39cd3924d79a8d96e : SUCCESS in 24m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9c7da89f59c54ebbb424f582d329b1ac : SUCCESS in 1h 16m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2a222101cdd844efa7dfb2f5bc250a90 : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/79351c1cbe33439bbc1178473557a571 : SUCCESS in 1h 25m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2e6f2df759d64a319221d873ec8173ba : SUCCESS in 35m 33s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8ba95cd7e6a8423eaaaa551b15c80c5f : SUCCESS in 1h 50m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/887d9f0bb23a44baa1991ff651256a46 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/80d76e000ef241afb74987753eb21ddc : SUCCESS in 16m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff432887758a4ccb9321af60190876ca : SUCCESS in 56m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3f9a23196bb4f2599c6c0f71ffa3740 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7c11c94ab746f298fa289d9e1433d4 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c233d9b585d64d2fbad9191318400a66 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0534e8a64a6d498aa5c7fd4fdb015f6b : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f64c7baf44418084a375aed83c0e25 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/13dc638e4fd748ec8cc6172050c7129a : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42f202bfbe7d42c0b9ff268389be35e9 : SUCCESS in 20m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9bed866109124910b823129fed08bb26 : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a1a1fcab83942e5a9d572ac97b580e2 : SUCCESS in 1h 34m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b2a7a9d6e9b454b8681da8a6d1e01ba : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/683834959b9944f9b1266d43d945d4b8 : SUCCESS in 54m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"976abe42872565b8a872a993a7bed270e0025180","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-24 09:06:49.000000000","message":"Patch Set 1: Code-Review+2\n\nfor contents but let\u0027s wait for k-a side before approving","accounts_in_message":[],"_revision_number":1},{"id":"87757fe3db332e535176e3bc8182a26831c693e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-27 09:27:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3604e6477984143c8c7f765a39bbecd7f6b9d6d3","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-08-19 20:03:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5a1e15daf5ede3deabb165c0567f16ee045d7a40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 20:38:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f534f36591c4993b3f69d60ab3e0b9f : RETRY_LIMIT in 3m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/532b97f597ed4b11baa33028356fcf3d : RETRY_LIMIT in 2m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fca6a5ad9e30457a81661d25967e77e4 : RETRY_LIMIT in 2m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd149864208448098204e233025830db : RETRY_LIMIT in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"089d4967e197d7f3a3ed4c569441ffa148c0d171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 22:01:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8b8912cbabdb4d90959a555817d834ba : SUCCESS in 19m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c40026722ab04ac69fdf371a401a6be8 : SUCCESS in 1h 14m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4e19d7550d764c5d800aeb2babf3bc49 : SUCCESS in 1h 55m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d385ac6f50d24facb76114d9d96431da : SUCCESS in 1h 53m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/392731aa80b245579bd6afd4e07821d6 : SUCCESS in 1h 22m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0b5967c7b5284409b0ba99a28b10ceed : SUCCESS in 41m 20s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/12df4a5412b44318b3433cc92dc788e4 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fcfeb342514e4b61a6f5fbde60cbaaa6 : SUCCESS in 15m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eae09e56056a41ae9fa2a68d3c849d14 : SUCCESS in 19m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8d0d1b210fd046b4ac5a3409052d7435 : SUCCESS in 1h 01m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f8c818bb39b4434a1893926a800d243 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd8ee68af8a436cbde3db0b58c2c790 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c35a908c1e4170b96930888ad0e411 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a315e3258044566ac89ee3ad45aac78 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea25db036844d2babced07e036c3351 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdfd38600b4c4cdda9daf41c4e18a704 : SUCCESS in 16m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e8b699b616c4c29955e1ef001be1f47 : SUCCESS in 23m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f1f619b70c843389e39fd0b1a113499 : SUCCESS in 1h 06m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d460b57f097f47bfbf772d01b935155a : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d2550639c794da68c3c537d5bec4e80 : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/82beb68d65e9482ca35392ffb8ac0706 : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8a0f3adf0aa00170e90cf300e3f7cc5a7da205a","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-08-19 23:11:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"048672af91cf5646e537be514fb94b4104ce1c52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 23:54:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4bae3f80a074052a99a683cfca44542 : RETRY_LIMIT in 3m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/13a99f8db05d46f4b2b48c136132e5c6 : RETRY_LIMIT in 2m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ad23e26e25a4e5f9666bf63260580cc : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d82cd498bffd4e6dbcc36974f32abf60 : RETRY_LIMIT in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12671e0f294a1e7737a24c5edc3484dcc70cdbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:32:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/af0b5e5dcc1e4f43846bc3287872e8c3 : SUCCESS in 46m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7ee1d62a4e2848bf8a90a2a8bb7be88c : SUCCESS in 21m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/89845917507a47a6af9671173467521f : SUCCESS in 1h 01m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/977ed38135ad465a8d61afd6d5d95586 : SUCCESS in 1h 12m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4356730580b84abfb473a4715205fd7c : SUCCESS in 1h 18m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5cd4d94040b34758b2382179c4e55dda : SUCCESS in 46m 25s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/3342f0e9032f4b52ba85919fa3b3a5ec : SUCCESS in 1h 04m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3f2275d70b38459cb7fd24f68d3fb452 : SUCCESS in 16m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/511dadb63b3044b5ad3a1144e5af032b : SUCCESS in 18m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f91896b7bac432a97b306283334077f : SUCCESS in 53m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8500b5dbeb0b48498ad801d24c1105c4 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aee8948a76a45f5a648c28c9c6a0ddf : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39b6f340aab84e99bddd2f9df86d0b21 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1ab59d0e654156abf4dd3de953455e : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9729975795f46258af2c54e55131382 : SUCCESS in 7m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3c1e1cbe963495cb0cc0d90f1a90aa4 : SUCCESS in 23m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38197ad9a772475f8a5fce5bfe5258ee : SUCCESS in 45m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b318e73eec1847eb8fc1e3cd76088be2 : SUCCESS in 1h 04m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/479d689513354e109344d9381bf5b80b : SUCCESS in 52m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50c299b33c2c4d01b0e8174b1d916a48 : SUCCESS in 1h 02m 56s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ebe874175f7c43bd9194d86f6e4b0bd1 : SUCCESS in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd1a7482be3b54cda4c3b1234ed7743939704ed4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-24 07:29:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5673a26dfd5cca7e7100580cdcf8c9c48af72b67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-24 07:45:57.000000000","message":"Patch Set 3:\n\ngtg; k-a in https://review.opendev.org/746606","accounts_in_message":[],"_revision_number":3},{"id":"d557afa2f2c2e0b69cab129b93ba50fdc9c794af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 08:23:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ad3f7ae3558a4b3d1e732cc7b50350a8edccce4e","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-24 15:20:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"01aab4d67bc3603d041ce1a1ccef2b683f2f6e4b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 07:20:32.000000000","message":"Patch Set 3: Workflow+1\n\nthe gates are GREEEEEN","accounts_in_message":[],"_revision_number":3},{"id":"67c233d527e794b6b72c9f076bca49b2b961dcf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 07:20:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da4b6f6fdeb35342bdadf77f97bc9fc38edf9f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 08:50:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/baa1e4e9bf854bf293233ef765a8da07 : SUCCESS in 21m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c644ab383972493883b164656e59c04e : SUCCESS in 1h 02m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/11b9ee7be5f24ead8c6cb1c7bf32360e : SUCCESS in 1h 07m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58592b3b216345cbb890d3ce901facfd : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dcd5f9f61244b2cbbce54dfb41bc1c0 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9bcbc5f055746f281b6db0790f62c1f : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/293605555a17449ebb22a3555fd166f8 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a6b0c251774cdfb7070f73364d0bd1 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae0398f2d6ad497894a5f9677cf84ad3 : SUCCESS in 19m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6657d94b2ad4047bc884a48a831bb56 : SUCCESS in 52m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/953274892f8e4bc6a19d3f81f911535e : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0712fdf320cc9e0fe8513a0a249502a774938304","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 08:50:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1953484b109106b09c332f5b184866f11586a5ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 08:52:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be8d0e14ff944368ab661d96152ee4e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d780694d65d996aba01d410857ca3217dbf21f27","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:01:11.000000000","message":"Created a revert of this change as Iaa98a8c85d7f105d701d0abfeeaeb8700828da29","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9a02633fafc6ffad03c45c62f27bf8059fd430f4","revisions":{"633e44e00527f77232c348fe1e755579d3e87fae":{"kind":"REWORK","_number":1,"created":"2020-07-24 01:55:46.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/76/742776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/742776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/742776/1"}}},"commit":{"parents":[{"commit":"1781af8b46b103bc47c57a50f7992240a2dd7bad","subject":"Remove neutron-server-opendaylight image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1781af8b46b103bc47c57a50f7992240a2dd7bad"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-07-24 01:55:33.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-07-24 01:55:33.000000000","tz":-420},"subject":"Enable mod_wsgi in Ironic API and Inspector container","message":"Enable mod_wsgi in Ironic API and Inspector container\n\nThis patch modifies the Dockerfile for both the Ironic API and\nInspector so that the \u0027root\u0027 user executes the container setup scripts.\nThis enables the container httpd configuration script to execute.\n\nChange-Id: I02b58ac571cd4eb3b9d4a814fe0cb907a3564e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/633e44e00527f77232c348fe1e755579d3e87fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/633e44e00527f77232c348fe1e755579d3e87fae"}]},"branch":"refs/heads/master"},"5433957d25ad131e2bbc74713875c5942adf7e8d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-19 20:03:42.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/76/742776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/742776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/742776/2"}}},"commit":{"parents":[{"commit":"e43ec971bc7a4ae42ddeee43248254cf7c26d3f4","subject":"[doc]remove sensu from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e43ec971bc7a4ae42ddeee43248254cf7c26d3f4"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-07-24 01:55:33.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-08-19 20:03:42.000000000","tz":0},"subject":"Enable mod_wsgi in Ironic API and Inspector container","message":"Enable mod_wsgi in Ironic API and Inspector container\n\nThis patch modifies the Dockerfile for both the Ironic API and\nInspector so that the \u0027root\u0027 user executes the container setup scripts.\nThis enables the container httpd configuration script to execute.\n\nChange-Id: I02b58ac571cd4eb3b9d4a814fe0cb907a3564e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5433957d25ad131e2bbc74713875c5942adf7e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5433957d25ad131e2bbc74713875c5942adf7e8d"}]},"branch":"refs/heads/master"},"9a02633fafc6ffad03c45c62f27bf8059fd430f4":{"kind":"REWORK","_number":3,"created":"2020-08-19 23:11:53.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/76/742776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/742776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/742776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/742776/3"}}},"commit":{"parents":[{"commit":"e43ec971bc7a4ae42ddeee43248254cf7c26d3f4","subject":"[doc]remove sensu from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e43ec971bc7a4ae42ddeee43248254cf7c26d3f4"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-07-24 01:55:33.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-08-19 23:11:46.000000000","tz":-420},"subject":"Enable mod_wsgi in Ironic API and Inspector container","message":"Enable mod_wsgi in Ironic API and Inspector container\n\nThis patch modifies the Dockerfile for both the Ironic API and\nInspector so that the \u0027root\u0027 user executes the container setup scripts.\nThis enables the container httpd configuration script to execute.\n\nChange-Id: I02b58ac571cd4eb3b9d4a814fe0cb907a3564e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a02633fafc6ffad03c45c62f27bf8059fd430f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a02633fafc6ffad03c45c62f27bf8059fd430f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
