)]}'
{"id":"openstack%2Fkolla~743178","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ia980be14fdf9c4fc15bb379b940ae491de035725","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ia980be14fdf9c4fc15bb379b940ae491de035725","subject":"Fix masakari-monitors image on CentOS 8","status":"MERGED","created":"2020-07-27 09:41:03.000000000","updated":"2020-07-28 16:37:24.000000000","submitted":"2020-07-28 16:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743178-1595954120492-8e31e638","meta_rev_id":"295555e85f0e78cf406d85f0a549adfaab3b9d8c","_number":743178,"virtual_id_number":743178,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 16:35:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-07-28 14:34:45.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-07-28 12:40:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-07-28 14:34:45.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 12:40:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-28 14:34:45.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-07-28 16:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90162ed1fcde93783ecea5eaf2969d2a0f2c53fb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-27 09:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d174e1fe6c0ba97ade1589dca384b3e02ea8dba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 11:49:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/64af9556eb7340648cfa33e34a388249 : SUCCESS in 30m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/700cf18fbda448c2becbc5399ec3c829 : SUCCESS in 1h 08m 08s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/458b947e380a45b9b45a0aafbbca2df2 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/936f3c1b39e94edabc92a4277351b24a : SUCCESS in 37m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8381fa2cbeb24ed69efb82aa0a5a4069 : SUCCESS in 1h 30m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d4113a72a15f4f69a8c51eb91710f2ef : SUCCESS in 1h 06m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/12733303c24e413d8833529d13c48da7 : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6940bf0842144e2c8307d36c0cc4f84c : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/768d8fa4b887471295c42ca20f92d372 : SUCCESS in 1h 50m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9d1c43df093a43949618b2e7578492f1 : SUCCESS in 56m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a465d4efbdf042acaee59cfb9716333f : SUCCESS in 49m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/141528300a0541cf8701a93094b9a986 : SUCCESS in 17m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4cd6cd9ccaa4882b3afcf05e4995a6c : SUCCESS in 1h 04m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be8cbc25e3c14c80a1a1a289980c9de8 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f48a59100d4f55b325c7558d4b097d : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf4e29c6f0474939a805fb1ae542f9cb : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0e74d872864191b1dda516b27d4134 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9814e9cda86347e0b082be215d5311bd : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e66e0f95f84182a658c8552429ea58 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6156f20aa89b4d1eb11d9e11b06dee14 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb8f9f3315984492975c1b80c28a0e34 : SUCCESS in 23m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5be066f2998c41f1baafd908a9ebba71 : SUCCESS in 29m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab6aa74a64a5486a833d1968d1293cbe : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/643e9e2d4ede4ed39886be4b64b61c7b : SUCCESS in 1h 05m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58295013cf3345619a9fd8a606c87da2 : SUCCESS in 1h 08m 54s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1d74a54905b887d2ce3ce9131b3e33fdbc3c81a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-28 12:40:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41fa267e8c8f40957670286a88e6dd4378e53d7d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-07-28 14:34:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b62bed95b3865fe48c32b821190d684830e1fcca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:34:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7476a6ae7e60f60a6ffc2ebbf878b2cd4c8c84bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/904a09729cb44a49b40f69e10b0fd044 : SUCCESS in 52m 44s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c986292d020049a6a5808cef16bb5ee5 : SUCCESS in 33m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1ca857187c6440429a609021aa41114d : SUCCESS in 1h 32m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80d0c9507e7348d79f0407bbf9c207a9 : SUCCESS in 1h 13m 02s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9c1a42093a6948dd8e06e6a6bc36f403 : SUCCESS in 1h 47m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a80e4681ef69464689765946b6a23a06 : SUCCESS in 1h 00m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/859bfcbb22424f6eb19d9e1280b76782 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b861a26cb54e4c8bef667a3dad125e : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30d86b84dcd343ab9e106936e4bdded9 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c855a6b6934d4a5492687c4c88b7e15c : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb431129a73340439c71655cdbab8411 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7913e6706a8b420580cb7aecc22f8127 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d2a91c8eae4e35b108c48f9613e9a9 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/96a4b88ac00d4af2963d96b3b65b26f3 : SUCCESS in 29m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cfa85f6126d49969399ee8ab094fcb4 : SUCCESS in 1h 01m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/84b5d4f9041d4eb48f56e711b4d674ce : SUCCESS in 46m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"ca16d9ec94d8a02c8e33ba0dc89ced22f6dd4d1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:35:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"295555e85f0e78cf406d85f0a549adfaab3b9d8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d237bc6a7943b69fe03923b4ce8c50 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87b6a1b4440249218540e8164561d1eb : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3e2da2e92302a7ba2eca2bc394529b0eeb24358","revisions":{"e3e2da2e92302a7ba2eca2bc394529b0eeb24358":{"kind":"REWORK","_number":1,"created":"2020-07-27 09:41:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/743178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/743178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/743178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/743178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/743178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/743178/1"}}},"commit":{"parents":[{"commit":"102ae176bfaebf10c041bb5fbaa8fa62215a5801","subject":"Fix Logstash 6 log rotation - Stable/Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/102ae176bfaebf10c041bb5fbaa8fa62215a5801"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-27 09:29:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-27 09:40:46.000000000","tz":60},"subject":"Fix masakari-monitors image on CentOS 8","message":"Fix masakari-monitors image on CentOS 8\n\nTrain only.\n\nThe version of libvirt-python allowed by upper-constraints in Train\n(5.7.0) does not support CentOS 8. This prevented the masakari-monitors\nimage from building.\n\nThis change modifies the upper constraints file in the masakari-monitors\nimage to use 5.10.0, the earliest version that does support CentOS 8.\n\nChange-Id: Ia980be14fdf9c4fc15bb379b940ae491de035725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e2da2e92302a7ba2eca2bc394529b0eeb24358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e2da2e92302a7ba2eca2bc394529b0eeb24358"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
