)]}'
{"id":"openstack%2Fkolla~743922","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2ecd3b029b956e000feaaa9e10a169521ba53d66","project":"openstack/kolla","branch":"stable/train","topic":"bug/1889065","hashtags":[],"change_id":"I2ecd3b029b956e000feaaa9e10a169521ba53d66","subject":"Upgrade Fluentd Monasca plugin","status":"MERGED","created":"2020-07-30 07:46:42.000000000","updated":"2020-09-04 10:30:48.000000000","submitted":"2020-09-04 10:29:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743922-1599215347268-8a9bf4a0","meta_rev_id":"cca26db862108662b02abf2311c502137078155b","_number":743922,"virtual_id_number":743922,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 10:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-09-03 18:12:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-03 18:12:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-03 18:12:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 09:25:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-03 18:12:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-04 10:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba8226488e2f2d8064c7c616b7aafb7f877f4180","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-30 07:46:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32ff46d8d5019cec3e9bb849585e2090cf29f769","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-30 07:47:10.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d03a95bae277ee7334ed94338bfc5d81880380a3","accounts_in_message":[],"_revision_number":1},{"id":"da12afb3d15e0bdf5ce63ef5d1fe86fc653c3793","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-30 07:48:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"173f191f1ad3750b0214beb0299afce1e18e9891","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-30 09:25:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1a8581a1d89dfe25fac7cbb1a0f1105f9bf0baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 10:19:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b32a38695c584741b11893ff00ee026b : SUCCESS in 45m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eb84a3db98804c6688fe2f0b8f0fd7e7 : FAILURE in 56m 41s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/43e999e1c1864ec786059de9b36b657a : SUCCESS in 24m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/48723adcf0c74c91befcce0a2aab4a1c : SUCCESS in 2h 22m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fab9f7f089cd473da324d73ef68ad41b : SUCCESS in 1h 30m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/039e30c6902347e99811b432e9a7c404 : SUCCESS in 1h 04m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2d757dd15e1e4b23ba5457b22967b3c1 : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b009a69de79f4be4b9c8692554a5f84f : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c00dbd5e81c24f7f9fa0b09b1ab482fe : SUCCESS in 1h 46m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5b20b93b49944bdd87b6a8deff7b04a8 : SUCCESS in 53m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6faee0562164437e84d38cf7705cc7b0 : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3278746e1c594d75b44d376d11313091 : SUCCESS in 20m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbaaefda40b04ce496653219202ccdba : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cfd3832a16c42348a2ce2851a655799 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b5ff01920a4843b3f4951b35c2bb0e : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a5889ced83d49af8357abe677d33ce4 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426a8e8f4c9a4ac5b583cb990d98537c : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eed8b4c231724da991ef31a55c1e3fed : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c54dd144a3544db881563e5d46aa8e9 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd77032e1324a029dd947608d40ace0 : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b2f82c504cc4c089025014da2e4c4be : SUCCESS in 46m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/201607bab2ee47778000cc3fbbcda781 : SUCCESS in 2h 13m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/124058724b734487a8b09582fb915d0d : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68b406e259314128976b70e1f3c2b7ee : SUCCESS in 53m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e040a79f2c94af493364172cbd75b21 : SUCCESS in 1h 03m 28s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"138b4c47e132d3170ea37a6d4ece9c71fa04b5dd","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-30 14:59:05.000000000","message":"Patch Set 1:\n\nrecheck\n\nINFO:kolla.common.utils.bifrost-base:  File \"/usr/lib64/python2.7/posixpath.py\", line 80, in join\nINFO:kolla.common.utils.bifrost-base:    path +\u003d \u0027/\u0027 + b\nINFO:kolla.common.utils.bifrost-base:UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc3 in position 13: ordinal not in range(128)\n\nSmells like something might need pinning, will investigate further if it happens again.","accounts_in_message":[],"_revision_number":1},{"id":"361d0829f959d0df921b9ff83de70f2958a514eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:51:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/74464ba8e2eb448f898cac15f858e04b : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/992f3963dae14358a8105dc6a3a8f031 : FAILURE in 54m 36s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/662af1b9955f43a6a8e33bec4b414725 : SUCCESS in 19m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6b9bf7febc784c538db785aa55e0e124 : SUCCESS in 36m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/709fb930ace6407c8adb12a696c87c3f : SUCCESS in 1h 10m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8e717dc530a3488aaa3a35b3185dc28a : SUCCESS in 1h 06m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6a7c6edbcfaa4e91af2229c51c56bacc : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/15225d8b95b74efb8b792579dba01e04 : SUCCESS in 1h 32m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/62e3990abe96429b8bfcd804dcc861de : SUCCESS in 1h 24m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fe32ec1776b9418fa81ee4b4cfe70ffa : SUCCESS in 1h 01m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe418345a644453f89385c5343080727 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/32851b9a4cbe48aca94c91f5cbd2557e : SUCCESS in 19m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37988ea561ad4f7ea69febd1a1c54ad6 : SUCCESS in 1h 05m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3141bcd480414b31a4449e8300e72789 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e666283c6e41f2904fd6f9401c5464 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87085a43bf204f43aeb3d878705dff53 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28baf26058a64d10a0a3f78c9b0ca194 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b62a6d68e2a04b4b841d54281e408bd3 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0787b5f239eb4290b4b947718634f841 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7358b6a299724bd399419edf7406b34e : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14b41e37d12a4bd594cffa9e46a4c448 : SUCCESS in 48m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a701ab1ab5f453fb179561b54f70f47 : SUCCESS in 33m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ebd8a328e9b4d38a2eb1ba06652ea3a : SUCCESS in 55m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/322793ebc5424a4ab8dc8caa826f31fc : SUCCESS in 59m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02e833da2d1f463fa9ca5e54f5841bcd : SUCCESS in 1h 29m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"634592ae2a8701abfce11a8457c3da55a206017d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-03 09:49:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bccdbb498ad2c579cceb38156b33d8cb82497621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:08:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed79c9705b034a049c54c03bcd8cca36 : SUCCESS in 36m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9f31435759924342a6633a96d91f21c8 : FAILURE in 1h 02m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2924863e932d44959af2ec99e602f9e4 : SUCCESS in 1h 08m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c2332e1c3d0c4eb5a043ab378bdf2967 : SUCCESS in 28m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2e381dd98cb24e73a952cf97eeb405d6 : TIMED_OUT in 2h 01m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bee7a1783e04484c9a8fb315102ce394 : SUCCESS in 1h 16m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d105f61c112f4324981b8fae543ef0a2 : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/08d312308b1941a682433887ff6ecddf : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/91ee7ac674f54c9ba1cc5b7fb0df1836 : SUCCESS in 2h 10m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/feb8d7e2471f4551b3e9751f368ca6ef : SUCCESS in 55m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e382f1d4027b44829d4927ea94046208 : SUCCESS in 24m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff0ddc3489a84284bd07aa115ca70d08 : SUCCESS in 15m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/099418c15b3c4c6daf6c4a0c1c9413cb : SUCCESS in 51m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9226efe5871e49eb82ba3ad59e6a3381 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86933a3f66da46568123c137f91596dc : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3d31393b9bf46c68662e282544c7e90 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3da322a6bdf1461f9b9422dcc3e8d8e2 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cf23e2ef58240e9a6c6e0774699bd50 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82aa5b1298534209a7748b00392450ed : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef14a76bccc4bbd85db20b36a6adaf2 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca79a79e0a194db984c207f3997cd0c2 : SUCCESS in 31m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e62cc72d1f549d8b4d67f1168aa408b : SUCCESS in 24m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cc938bd2fe34063b3e034b737dd6945 : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e0b0a85b79146cca8213cdbfbdc9aeb : SUCCESS in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ca7d5805dfd4cf2867e17b54787f4d7 : SUCCESS in 47m 29s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"36d0542efd94f7789b1044ff01bd71f24942402a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-08-07 09:28:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a7f03c05cdb1c12a13c6b566a5669209927836b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:25:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1dd953f5c97b483ebd94c0c9c78e7998 : SUCCESS in 49m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ddff159b5b274b9f92a8c5f9ddb32d7e : FAILURE in 33m 14s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a79a37b16df742f69310ba1fb79eeeb2 : SUCCESS in 1h 03m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cb9422c112eb4316ab966ad9fe8f8366 : SUCCESS in 45m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/87aacd480ee74b54a365cb5f03b14e01 : SUCCESS in 1h 19m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0aa39271e97f436aaf89b50fab5e9363 : SUCCESS in 1h 21m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1736607ed28449bd80c587fdaf3bea2c : SUCCESS in 1h 22m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/30a13814502b4054901eb46de9b1d632 : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4dd4080da847459da6eecd20cfbc787f : SUCCESS in 1h 55m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9204f64031214daa992f6fb7adfe341a : SUCCESS in 55m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bdc55cdd9b1c49bd9433ac01b37c8f69 : SUCCESS in 38m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aa764fab2cb9405c8554f76a5a577c8d : SUCCESS in 26m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/497b10da783945559ac2a08d04ce105a : SUCCESS in 58m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66035e7f794549c1b294f36077f9b6c3 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfae433e074444bfac185bcdec8a63c3 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74d4b798d9b644cdbe45c354a4d6050b : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6726eb89b2d94a47854aef37cc42f01c : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f733a474aeb477e9de2f56d831995ab : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d9dcc48d41461fbde27b38c5ffcaef : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86781850df374e61a0c1ec1698d06850 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a4293c6380e46f9bb060209aeb840c8 : SUCCESS in 22m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76dcd30ca4d04990a3f0f60938440d1f : SUCCESS in 37m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a79d6c962b04e379c5a19ca8ab3f2d4 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/46f29794fe0e431c89028b7b8b386bd0 : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57f1a25a936c4a84aa1c5b76c902acaa : SUCCESS in 1h 08m 05s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"91c5a155b9e74ea0a3256f045dae807e382efb33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 16:54:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5d6f883cf0f43c9c67113ddd99b3854f05c84f57","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:12:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5356df6a83b1c88549d58ba7dc66dc89daf3c0ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:15:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"65502bf8b207775d7d0b5f37924a740b595702ab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:15:44.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cb6d41223dbfc5193953cb33078e7304edb4a073","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:18:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ad46c8e66d333ceaf04699338a67f2cdddd85c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:13:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/542da8844363462d81f6a4587d69af51 : SUCCESS in 55m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/58dba285992d4890b73f4a62a3390ae4 : SUCCESS in 52m 45s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/32e6f2c10f6245978d65dfdddde1e72c : SUCCESS in 24m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e3cf15a9bee24660bdc1ecf766a5c599 : SUCCESS in 49m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3c7348d63a3c46b090d01f7c68092a10 : SUCCESS in 1h 14m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fa455c42edb24393a344f720cf00ef23 : SUCCESS in 1h 08m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fc8a5c032c844df88d17d95e94541096 : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f1df52451cde4b2bb3f62391b82fb517 : SUCCESS in 1h 03m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9dd18410dd994e5dad3474bfe94fb197 : SUCCESS in 1h 35m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a26680a111fd457bb31924d122d6a903 : SUCCESS in 50m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe78f03d753642af9c53f2cce5ceb216 : SUCCESS in 22m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/61e18d5898264a559f3b433f46d66395 : SUCCESS in 22m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a589bab7c0b94e2b9ec39ead72098b2d : SUCCESS in 51m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df5091fb3a2a4343a95e1213a4e111f6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bc02a3d3ff41ab986dda746532aa04 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b3e998718854d259b7b7eba4ff64a50 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24830310ca3a40148d6513814fb80a05 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cef6f7bf0b448989e9b8a1926d48d65 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f955e6b83a497e9ce66dc0352aefd4 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ad0039e65864a489ffaac3eeb127321 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e8ca3e2bb594a66976bfcad7c6937d5 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1b1cb252a66448680a874eac48a34a6 : SUCCESS in 26m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10c4e55f4fcc4a36bf8fcfa7d10292fd : SUCCESS in 51m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d49046acf2c44767bff377535292bbd6 : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2777176fce24a7d84d00cccedf4b692 : SUCCESS in 1h 01m 54s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"3aeeaefbe156df7bebf3101ff680d80d2b05fd6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 07:02:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"658ea9d08146b5b81118fc2fdf90408c692ba935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:28:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b4738e8bbdd649099c50624980b4c822 : SUCCESS in 1h 18m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/408ff19a4b6a41c99e68a244ac2f2658 : SUCCESS in 1h 08m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c87ab4e24b09487ba94ff5310f908f2d : SUCCESS in 23m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/58fd766663bb4b8f957594e0f5dda9a6 : SUCCESS in 29m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b8676ea2afc14d9ebfb2e883d91dd790 : SUCCESS in 1h 22m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5dafb4ab7b5343ad8697bc4e5820da35 : SUCCESS in 1h 00m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d076800d41b74cb09b47c52d8473a986 : SUCCESS in 1h 14m 35s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4bf94d65e94f4e7fb09e7661b50d14b9 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e366b10e4cb5465a96f2dec53a3f5e19 : SUCCESS in 1h 45m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5dfece72d9c94b4d8c20a150c8a62871 : SUCCESS in 1h 04m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b7737d2691b042f998a6d7ee5608984b : SUCCESS in 58m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/941f75d86e91476db982e57ec43bd5f2 : SUCCESS in 18m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/04da774144434ecda56225dc77c3e1d9 : SUCCESS in 50m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb00153ef2f4d84b61f583406806c96 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad71b59b64f41fda2dc34f49c504e77 : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bac756ad200d4e09bc700345ee187b94 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b6e31d5c7b44198838cf497855ec889 : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/324c40b3f4d5460291e103ffe0417714 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447cf5d3c653417a828eaa0eb6efd348 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5000634f3842c782ee9b6e418f10cf : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73aac26a1a6e4dba92847ddc7e4a63b0 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da9a0cbdb9ec47ceab76775a06989428 : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9fc92234e8f41799b416a61aac29813 : SUCCESS in 49m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ceb1791aa91046819bab678a9fb73eb0 : SUCCESS in 53m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b6e2508671ac444786a655e66477d8de : SUCCESS in 59m 36s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"d9b30fd8b9cf4189b4b6f8f6bb5e0f995f522e1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:57:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c8a73fd9cf6075a59d90708c5a442edc633fd109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 13:00:03.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"43ad753d055803dffc60c035b6c64810d6457b57","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:47.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"1fdeb2577d90df783030b1b6aba1270ad7f20886","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:51.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4b64da96986897ef3cf9fcc72448edfab0649ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:15:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/322533f4bdab4515be9a4c1b5e3d81a5 : SUCCESS in 33m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/511dc11e3dfc46cf950c3b0ec9e68312 : SUCCESS in 48m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/002b2ddb12744e12a24005a0eff71fce : SUCCESS in 25m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ef9e018af5ad474f965edfa75663c822 : SUCCESS in 30m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a12371b251a2494188fedf4863b56575 : SUCCESS in 1h 17m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7e00c4e7317746799b2400fda15c3d30 : SUCCESS in 1h 53m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8a079de748f646689dfc53c24e2cea69 : SUCCESS in 1h 22m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b58209859a744657a328793eeb23a9d7 : SUCCESS in 1h 04m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7f46ee834e74031836252a7e2ec1245 : SUCCESS in 1h 24m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6e82d5f520534b008d18ae7554c52be5 : SUCCESS in 1h 06m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33e2e5a994d343ba8b4f67ee178e541c : SUCCESS in 21m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/432345ab288f45919b4f34beab051b0c : SUCCESS in 17m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0ee7806642794488b6bcde8c65d834b0 : SUCCESS in 53m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8bab1e58df14403b0fe6d36605c82d1 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80377247346d4682a4a751c86caa9d6e : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/061255534107457e84390955d129bdae : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b716ac0dfe4ceb8d7af0eb7c5465b2 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/574ed4ca01624290af12aa5f0ad1ee34 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cf6ab9e8b14eb8b708cc647da4ee32 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d98dd17edd64547acf34af393a472c3 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2270b27dcca646b193c348566072e13f : SUCCESS in 54m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa9282c833264e64afbb01e5b508581d : SUCCESS in 23m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9284b99d93de4af3a35ec9218aef7fd1 : SUCCESS in 47m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5bbd9aee25664651a5ec4538b6bd705f : SUCCESS in 52m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/673a9b4b023b4e07b54432d3688ac18b : SUCCESS in 1h 11m 24s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"96d494e954134db0c6a5f3ea14d03580f9b44ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:54:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ca8867d90fecfa8f00c97959874671ae5ac996d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:37:12.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/485dfd1f4a5d4bcb8e825e6d2e78170f : SUCCESS in 42m 59s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b6250dc2eda445fead6640b8e0be65ba : SUCCESS in 26m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ea12eb4c80d548f0951dd62fe77bb4e2 : SUCCESS in 1h 09m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/792c68e4a5b944b896e3ca221b774930 : SUCCESS in 1h 03m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6b6ce04e425c44aea0f20fd05ff39034 : SUCCESS in 1h 20m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/273eb7ea016349d39b4aa88162b9e742 : SUCCESS in 57m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b4060fff3fa4c309ed6f9967aeec09c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a764f498d34c60a6b6223b02e619be : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/251424d264464e7fbcf5a38d00687b4e : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6e7856898b84da589b32ba538366e3e : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c675b02bcd404da09ddcb6bc32238f01 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b552837ab11d4f1ba4f893481d2920b2 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089f07da9baa4f439e1d33f3c72ffa3a : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d529147e15f446f6804573ee0f4e6af8 : SUCCESS in 23m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84866cb25e944c229b5a5a5da2d75d00 : SUCCESS in 45m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/455276a7be8743199285d0667a77ae69 : TIMED_OUT in 2h 33m 31s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"de40c8f27efe0dae6965d18465688a0e7b9a1b02","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:14:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4a873410827c4ba561df32f5125b7a043b5684dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:58:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e708d1a2eb9d46a5848ceff7e23f5a0b : SUCCESS in 56m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cd1da2b020e34e8fa3de046af6eb27e0 : SUCCESS in 40m 39s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f77cf9363b164a948c0cc4f06b6e120b : SUCCESS in 25m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa33132c4da644e399dbd29fbe80e1b6 : SUCCESS in 25m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/070dbd73aa18466e9012ab465d1d26f7 : SUCCESS in 1h 27m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fa0d6b7729bc4fd088a803adb9857814 : SUCCESS in 59m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/51548ad7b9d54281aac7a7873a736392 : SUCCESS in 1h 15m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2015e01ac315408abbd9622e1d3f3a62 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/57a026fccac945fbaeffb3d328bc3a33 : SUCCESS in 1h 25m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c710e02940f740c8b6fc04ae48eecd20 : SUCCESS in 47m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4df3c69b3619482788b92f9cffb15f82 : SUCCESS in 20m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6eb0dd7d39da40c0a1a036c808f0d62d : SUCCESS in 20m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/06b4bcd991294640bd38bb8889a38b0f : SUCCESS in 56m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83186c3a57db421bbe7e12c3afe4c83e : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a64947f2bef4d18826d8c2db1123a4d : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8a6204673a04cdeaa19cf4588ce819c : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/976ae3abe3614934b3fba9042b60ad48 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b02d241e6af440d2a61e6682ee829c52 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a175486a922c48b4bc952c51addc5831 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9538ed962cf4f939a9046417de5fb32 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9fa5d5163f54656a9bb04a4bb523b3a : SUCCESS in 25m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c788e33aa3cf4a1092186a8b4acb84ea : SUCCESS in 38m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/14fd99874e324196ba6019fca840c96b : SUCCESS in 50m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f59df47a711c4c7ebfb8c57e0f244b2f : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3912871679414fb3a427669711a7c9ae : SUCCESS in 1h 11m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"85155e3c40318cf444e577409ae703572f2e64da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:59:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1d6354d6a4bd02694becf4ca9fa648cb189924c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:29:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9ffd0d29ac8afc374cc28beaa5e646425f21f48e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:29:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d04bd5dec38f470aa44a80583273d5c2 : SUCCESS in 48m 25s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/004e05391b17486b8c82678d81f34ec7 : SUCCESS in 27m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f6d8ac9b88fc4fb98b577accdd58751c : SUCCESS in 1h 10m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a9d0d75c537d411f8fdefd66bb80a7d3 : SUCCESS in 57m 11s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/326e97885ca348a996d68df210c3ecf0 : SUCCESS in 1h 27m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a3c6ad5488bf48c1a99fc41c09fd8c95 : SUCCESS in 51m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c747db24503c471d8fb0b29050a38975 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428b778fa43e4b46b279bf2aa07fb4f2 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/951e9a388b2145a8af8a81dc3cf1e5f7 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ece8c2a15544352959dd9dc5eb21992 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5edb61d253e149229fd0f69d9856cd88 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c05402ce9e645b9b6df2c94b165c86d : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9913dba38e44a7b7c75b37f281fd4c : SUCCESS in 3m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/960c3d7072c94e3691684d68c6ef5ff4 : SUCCESS in 20m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cf74884b2414f02b4d69759467264a0 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4fd18c68e9c4554b327b4cb884aebea : SUCCESS in 1h 07m 00s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"30e244e934f5f9039012eda5c52dc10f9667c4d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:30:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ade18099a44405cb961c5fd5d2a4e81 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3acfbd9aa6e34219bdd730a11946d7bf : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bfc8ce085b338df91eabd55b833490e3ec2a02e2","revisions":{"8e59b5e1aa94a989c6db0c95d0aa6345746258d9":{"kind":"REWORK","_number":1,"created":"2020-07-30 07:46:42.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/22/743922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/1"}}},"commit":{"parents":[{"commit":"e3e2da2e92302a7ba2eca2bc394529b0eeb24358","subject":"Fix masakari-monitors image on CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e2da2e92302a7ba2eca2bc394529b0eeb24358"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-30 07:46:08.000000000","tz":60},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e59b5e1aa94a989c6db0c95d0aa6345746258d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e59b5e1aa94a989c6db0c95d0aa6345746258d9"}]},"branch":"refs/heads/stable/train"},"abd2712c6d426285bd3e24aff6a297ff221aa975":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-02 16:54:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/2"}}},"commit":{"parents":[{"commit":"f8208f6b5b330f59dd1aedcdc2cb9393a05a79f1","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8208f6b5b330f59dd1aedcdc2cb9393a05a79f1"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 16:54:52.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd2712c6d426285bd3e24aff6a297ff221aa975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd2712c6d426285bd3e24aff6a297ff221aa975"}]},"branch":"refs/heads/stable/train"},"5ff8d8a25cf04010eb226ac94f891aea68693c92":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-02 19:12:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/3"}}},"commit":{"parents":[{"commit":"e08c1882f5543d05e61a72d7b2f1c3aad988852f","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e08c1882f5543d05e61a72d7b2f1c3aad988852f"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:12:41.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ff8d8a25cf04010eb226ac94f891aea68693c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ff8d8a25cf04010eb226ac94f891aea68693c92"}]},"branch":"refs/heads/stable/train"},"4cb4827e5e5a20b185343095dae6f12e38d11189":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-02 19:15:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/4"}}},"commit":{"parents":[{"commit":"6dca929e2aca7f1bfb7ed003c869a4aac555c7e3","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dca929e2aca7f1bfb7ed003c869a4aac555c7e3"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:15:24.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cb4827e5e5a20b185343095dae6f12e38d11189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cb4827e5e5a20b185343095dae6f12e38d11189"}]},"branch":"refs/heads/stable/train"},"2bd584439f6f5c0cfc83f810b15cba951118a58e":{"kind":"NO_CHANGE","_number":5,"created":"2020-09-02 19:18:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/5"}}},"commit":{"parents":[{"commit":"e57a177243a532cfcdc3179929cdb01241aa78b9","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e57a177243a532cfcdc3179929cdb01241aa78b9"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:18:09.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bd584439f6f5c0cfc83f810b15cba951118a58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bd584439f6f5c0cfc83f810b15cba951118a58e"}]},"branch":"refs/heads/stable/train"},"65f66461b8e05734a8892d42e896d8d58997752e":{"kind":"NO_CHANGE","_number":6,"created":"2020-09-03 07:02:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/6"}}},"commit":{"parents":[{"commit":"07f101611b5415753c05c7bb45fb6a8438956094","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07f101611b5415753c05c7bb45fb6a8438956094"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 07:02:05.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65f66461b8e05734a8892d42e896d8d58997752e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65f66461b8e05734a8892d42e896d8d58997752e"}]},"branch":"refs/heads/stable/train"},"bfc8ce085b338df91eabd55b833490e3ec2a02e2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-03 18:12:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/743922/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/743922/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/743922/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/743922/7"}}},"commit":{"parents":[{"commit":"2ff242c4a270258fed0a0dbdbd3b66a402cadece","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ff242c4a270258fed0a0dbdbd3b66a402cadece"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 18:12:47.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc8ce085b338df91eabd55b833490e3ec2a02e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc8ce085b338df91eabd55b833490e3ec2a02e2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
