)]}'
{"id":"openstack%2Fkolla~743923","triplet_id":"openstack%2Fkolla~stable%2Fstein~I2ecd3b029b956e000feaaa9e10a169521ba53d66","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1889065-stable/stein","hashtags":[],"change_id":"I2ecd3b029b956e000feaaa9e10a169521ba53d66","subject":"Upgrade Fluentd Monasca plugin","status":"MERGED","created":"2020-07-30 07:47:10.000000000","updated":"2020-09-03 17:20:54.000000000","submitted":"2020-09-03 17:17:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743923-1599153459926-941d6ed0","meta_rev_id":"2b2d48760520d30ba3e892a50f1a44af19920445","_number":743923,"virtual_id_number":743923,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 17:17:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-09-03 14:38:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-03 14:38:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-03 14:38:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 09:24:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-03 14:38:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-03 17:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9089f95ef338502179e5c654aa1cf8007ae186a4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-30 07:47:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"7f263c64ca8962af97e7a145dfa5864b49c40910","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-30 07:48:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fb49f21ae08fdbe8e3cd57ecc8f05bffdc1eb01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-30 09:24:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f3d677c77df2144c4f207e2463a7f96125a14fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 10:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3c12821026fa45778b6c6aba04ce4a7e : SUCCESS in 38m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f8f14024da244331aceb64eb4fc35ab9 : SUCCESS in 2h 22m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8d8e63f19f4b4fba87a4f594fd3932cb : SUCCESS in 1h 08m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1e89a45a5807422fa006f444a205b9a8 : SUCCESS in 1h 04m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c390ae2cea424389a81c0abe151a8987 : SUCCESS in 1h 33m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ccb14ef66a9b4ed2b486d2349c02e0a1 : SUCCESS in 45m 29s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2d3e9517b15148f3a5c7d3b5d9700edb : SUCCESS in 37m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4bd90946fc4047ecb42fa7b1a47a82da : SUCCESS in 49m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/104c08ca883c40c088974b9d23e79127 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c19bd9eb55402b90515007ef70b6c8 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92da1a67e6db40bdba8365a96a01f37c : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432410fa1e374f16b1c279b6b5d235cf : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1deb4c5a9b0d409eb64144e66df7f8a1 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4552147c3d6d401bac8417fe7ba67f4d : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aa564078a5c45ed8ef7730490b26a8f : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b69cb05bf2884056abe4175dd81299d9 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/013a74e7aca949d7ba6e443af9acacce : SUCCESS in 26m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0d9ce5196c7422786616bdca4f791e1 : SUCCESS in 37m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/586664045bef44ec81ffe3161bf9bc26 : FAILURE in 37m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3bbd7f2b9d664f5396d2da431806f04d : SUCCESS in 1h 01m 47s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1c4b4596cf7958701113ee6a1594c3ef24f492e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-03 09:48:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7fc539edb8c0bd4a693183b64f1ba35c13efcb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 09:49:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9447c54b71248662bf47bc9cf39011fdb676d0dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 11:04:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/10d084d51ae04fd383e37896da087e24 : FAILURE in 56m 28s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e3e67770c34141c8bf271cf1f3bd2447 : FAILURE in 33m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f827195015414d2384455dedad5c1370 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6529223e3944100ba28fc033473e178 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c967fa3eff41ccbfb33a875d32d004 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66a0771a96384c2bb04a6c0185fe30a5 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3da32a11deaf49749bcac66822c00ab5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0dceb1905742649c4f34a73b651858 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ce1dc4c51e436ea8d003c1f25f0faa : SUCCESS in 3m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d6643cbb5b348a08c790b1525924714 : FAILURE in 19m 38s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"71063e1e8164150340ff7bc39db301c60b11aba9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-08-07 09:27:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b5ede1c9dcb518f14141d372d77c5b38c9c2d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:59:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5f40b6a76c40457f8199b2dde6aa28b2 : SUCCESS in 38m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a4b634ed8cb4c819454f587c8ee10e8 : FAILURE in 1h 22m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a3485b7ff6dd404c874275e4a13173f1 : SUCCESS in 46m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4c504cb3a76e434b97e41c916d8f9c1a : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7fde0d7d06243d880c4b4ab5cd218b3 : SUCCESS in 1h 30m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9d1d55a6dd8043bf962fa758532a1854 : SUCCESS in 1h 04m 29s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cd9ff7c8683b4c55bae114e94a322241 : SUCCESS in 49m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7db22c6f6c1d43dd945faa3d6c1a6dc8 : FAILURE in 1h 09m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e13896e2602e43f580eacc4148ac0ac1 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a57c6fd42054e1abd454a11db051231 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6fbfa244ddc4d739d413066a7e09870 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/079cb9d5fe2440cba41038533d55aa74 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0426f9734df743ce917f08ef21878346 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d5b85fe9cf4917817340dcd84c4a5a : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/345cd0fad127449fb6759736df726af2 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e24b95044046488c94349f3622d1923e : SUCCESS in 18m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/850c8802805144f1bf5d4957c0e4242e : FAILURE in 30m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad78c122d04f45d3998b2b73236c6855 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/151efd0aefe549608e8a34e118b266e7 : FAILURE in 35m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d0e8cedb84d4acdb17ffec6edb59378 : SUCCESS in 1h 03m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5fbd8840a3b6f0679f94e66b4aa09dca79a99bbb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 14:38:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"99a1c98f59debd1a9f2cb133830ae86f5f81d4f1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 14:38:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ee4db3d516637bc06ec9890623b4b00ade1cf5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:17:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3ede4945a10645b0b873ccb1e97c3c0e : SUCCESS in 1h 02m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2758b92eda984b298c488914828ff845 : SUCCESS in 1h 01m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/53084362f6d146c08ab874bfca7912c6 : SUCCESS in 1h 22m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/87ae17a6c1294730a02464dbff154473 : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/34e1bac4e1bd4660853e797a12da0e6b : SUCCESS in 1h 31m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/571c69aa4db543fd906cbad73f1850e0 : SUCCESS in 37m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cbe9fae7b6544fa4a3d58dd38134a76d : SUCCESS in 50m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/345aa9476acd43cf8e92b5d56e538680 : SUCCESS in 55m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0da02f5e662d4ef5981cfdb7a5c60021 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4675118cdcd14ae6a8c2953050701749 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac20d4cd40934f1ea23bef7910874c80 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43115a51c8434887af4b21d431999b09 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2baa4a807b8d424fb1ead7df902604e9 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa94b276842a4eb0b1d5fb7de4ada081 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ca5aeba0b1f42e9ba3978e17dd2f0f4 : SUCCESS in 9m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ee7c34b940343b1adbef4727deeab21 : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97fd279e94074bef99c9717c0ef179e4 : SUCCESS in 46m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b9865ccfea04a0ea7fc5ca65a86be0d : SUCCESS in 1h 02m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/888031f26a464ab9b32a19f0ca3f3451 : FAILURE in 36m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a713dade0ec947c5b2ae2a9ec1a5794c : SUCCESS in 57m 01s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"1ad096fc1f5240b450a1f9b061ea66aabba60704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:20:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5dc05adc0713227783453c69f182b0da81bcabae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:17:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6bbd0039e6364b09adb67e111a621d22 : SUCCESS in 50m 52s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fe053dafd8ef47c69c6b939da78f0ef4 : SUCCESS in 40m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f579513d680494094636515896cdba0 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb33e438a69b4527be4f5722cf171681 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/684690090b02456baf18f74e2c1629a6 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd9a864935b64e5e9fff6d162b8464e9 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29e3425850f241d4ba57fe5eeaae1d45 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae371bc02b234b8aa596efd24e963381 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c266632241924f7e9fa9f0c94ccce470 : SUCCESS in 3m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7235b3bbcfa0425c87f7cdeb877e45a4 : SUCCESS in 21m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"fe59a0fb95169b393b5e6e85b9fff43805a7fb08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:17:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a62780ec2895ffcffaf2e26e2bdbe32d59701381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:20:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba7602b59214618975d93f70b9dafc2 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3817d36fed714324ba32f68e2f729ba8 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5b5e926895d8833ebc2b54debd74a4d9d72492d","revisions":{"d03a95bae277ee7334ed94338bfc5d81880380a3":{"kind":"REWORK","_number":1,"created":"2020-07-30 07:47:10.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/23/743923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/743923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/743923/1"}}},"commit":{"parents":[{"commit":"7038e3ff03c6d08e7b794f9087675964ddbbed0a","subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7038e3ff03c6d08e7b794f9087675964ddbbed0a"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-30 07:47:10.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03a95bae277ee7334ed94338bfc5d81880380a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03a95bae277ee7334ed94338bfc5d81880380a3"}]},"branch":"refs/heads/stable/stein"},"e5b5e926895d8833ebc2b54debd74a4d9d72492d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 14:38:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/23/743923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/743923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/743923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/743923/2"}}},"commit":{"parents":[{"commit":"c8123bf6f385061ea7180efcf0d359adcea7a57e","subject":"[CI] Fix wheel url for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8123bf6f385061ea7180efcf0d359adcea7a57e"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-07-27 11:03:02.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 14:38:36.000000000","tz":0},"subject":"Upgrade Fluentd Monasca plugin","message":"Upgrade Fluentd Monasca plugin\n\nSwitch to YAJI JSON parser as suggested in [1].\n\n[1] https://docs.fluentd.org/quickstart/faq#i-got-encoding-error-inside-plugin-how-to-fix-it\n\nCloses-Bug: #1889065\nChange-Id: I2ecd3b029b956e000feaaa9e10a169521ba53d66\n(cherry picked from commit 556d4353aa846765557cfdd3c6e087a25248c517)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b5e926895d8833ebc2b54debd74a4d9d72492d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b5e926895d8833ebc2b54debd74a4d9d72492d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
