)]}'
{"id":"openstack%2Fkolla~744498","triplet_id":"openstack%2Fkolla~master~Iae240eae5d3e4a9efa42ec73940728775dfc25fb","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iae240eae5d3e4a9efa42ec73940728775dfc25fb","subject":"remove duplicate package(planned)","status":"MERGED","created":"2020-08-03 14:47:00.000000000","updated":"2020-08-04 12:27:16.000000000","submitted":"2020-08-04 12:25:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744498-1596543937857-c9d102ff","meta_rev_id":"c165eb7daefed49b8e6825f3d62f4962f7c13ec0","_number":744498,"virtual_id_number":744498,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 12:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-03 15:13:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-04 11:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-08-04 09:41:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-08-04 11:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 15:13:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 09:41:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-04 11:01:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-04 12:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e50af13013f3b4f821a5dbced4be1f7df45e01f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-03 14:47:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279f2f838422d065afa583d4701801bfa4803ecb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-03 15:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8cc38cc0dd7d433ebc8a18cb72caff87 : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebcd1684f77edd708b731f44a04e9904a2550e46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:26:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9083aff91c6d4704b24bf0225902834b : RETRY_LIMIT in 2m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba1fdd5f09bf4c249b4872376637118b : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c66934643e7d489a94fc2737f232ff2d : RETRY_LIMIT in 2m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/00c4c4be4fff4e56b476444655454ada : RETRY_LIMIT in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9fdbe27045741ab8c68964ee063150999448a49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:04:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bca3aeb49af141ed902260504cf6012a : SUCCESS in 21m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83ffed0cd3b8418f96dfbb79396f849b : SUCCESS in 31m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a4cd762c525244bd90a57e8a13653343 : SUCCESS in 51m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/150808ff95824c7cada58670b4a1bca3 : SUCCESS in 1h 08m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/18786df4c60845f18dda3fc0e456caf3 : SUCCESS in 1h 12m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/416c3a9365c748f8993c1c933bccea4c : SUCCESS in 27m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1c07283767f437899cbaa9b8a02e007 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/505c144036ef4e2f93830e73cc54e93c : SUCCESS in 15m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c08a495b6f1442f48818db4701de2b08 : SUCCESS in 1h 00m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68f1224610504e9d9719605406774d3a : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521f54c7327e4ddea504a982cd3609c6 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d27663e157944ca8c33eda8d9ce285a : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b7f3b3c17a74698942219267c9b8790 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effc488081ff4617b99cc4d3b90ebbbc : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fa5cc88606b4706be1b33cc82081d3d : SUCCESS in 21m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3959d20f6ed4425b96daf975eb4a8082 : SUCCESS in 18m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5dc3ec709a14b34932cb58fda17802e : SUCCESS in 1h 08m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cad3215776db4b239e8532f174b8c340 : SUCCESS in 56m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a151e57833c445ba6330304409e7793 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e06de426f59ae2835cb47225699019ac2477df2d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-04 09:41:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16496f59ef70898af263f0ef2a6ef6675546c7eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-04 11:01:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f749ce60a2dbc9361537bf2a2a924104d0478d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 11:01:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77238b8451e86f416ad4abe75dfdfadf95bafb82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:25:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6f0b0ad0468344b7a515bb5afdb8616a : SUCCESS in 25m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c6beef5f5e94434486d06fa6e317b670 : SUCCESS in 1h 08m 55s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/34427cdb63884bebb4f8a69baed66162 : SUCCESS in 1h 06m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f448029b685488ebc70c4e2f25ad95a : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe5ae27cdaf4a249e2bc00e7bb2805e : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d2ad456a6a4ba583b4c19350d6338a : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edc3515e23834a9297e98af251ea0f53 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5651d9864b6046c492becf80d91782b7 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92d1615fe92e4c79bb182faabc4b7f18 : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83215cec8e334ce895db52b16c5c36b6 : SUCCESS in 1h 01m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/547a438ce9824d3aa796b42f881529ba : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"af6c9ba269606e1102c86034aa79a36e37a9637a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:25:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c165eb7daefed49b8e6825f3d62f4962f7c13ec0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:27:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e0c6a72eeb49528841ba54142001ae : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ec824b0a192d06b6a71dda126e5d1b00b5c3238","revisions":{"4ec824b0a192d06b6a71dda126e5d1b00b5c3238":{"kind":"REWORK","_number":1,"created":"2020-08-03 14:47:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/98/744498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/744498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/744498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/744498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/744498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/744498/1"}}},"commit":{"parents":[{"commit":"69092eb574222564fcd3ce752a41db7fe51d61b8","subject":"Merge \"Upgrade Fluentd Monasca plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69092eb574222564fcd3ce752a41db7fe51d61b8"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-08-03 14:11:04.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-08-03 14:45:40.000000000","tz":480},"subject":"remove duplicate package(planned)","message":"remove duplicate package(planned)\n\nremove python3-pymysql from placement since it has added into\nopenstack-base\n\nChange-Id: Iae240eae5d3e4a9efa42ec73940728775dfc25fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ec824b0a192d06b6a71dda126e5d1b00b5c3238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ec824b0a192d06b6a71dda126e5d1b00b5c3238"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
