)]}'
{"id":"openstack%2Fkolla~745105","triplet_id":"openstack%2Fkolla~master~I8921befe51da4282121443849177a7ca5ebe8822","project":"openstack/kolla","branch":"master","topic":"bug/1890567","hashtags":[],"change_id":"I8921befe51da4282121443849177a7ca5ebe8822","subject":"Fix kolla_set_configs --check with a directory","status":"MERGED","created":"2020-08-06 10:02:41.000000000","updated":"2020-09-14 10:09:05.000000000","submitted":"2020-08-28 19:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745105-1598641422675-9749b533","meta_rev_id":"0842f8f41ad4c2399c29d75fe9a5b61813c00bfb","_number":745105,"virtual_id_number":745105,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 19:03:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-08-28 15:01:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-08-28 15:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-08-28 15:01:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-28 15:01:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 07:52:52.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-08-28 08:22:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-28 15:01:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-28 19:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bd60df730dafd07971dd61e35e49ffc08beae5c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-06 10:02:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37482ca9224adb8ad704d77b3181665f5ccb5573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 11:41:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d3060cf2574480d9d6caeca1a8d60f7 : SUCCESS in 57m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53e20d53b949462782eb19ab484258e3 : SUCCESS in 44m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/baa60a4edeea464b8a61ff31ffe21a84 : SUCCESS in 1h 34m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c650d3695d124f8dbaf5aecbfc196bef : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eb4473dcaf980b06621f5cd7dc5cbab3cdea6f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-06 12:40:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"101e5b69538b943396af8613f742165cd9bce922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:22:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7424776063a9438ab02a7c1229c65f14 : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab591120204c455bbc53fcf8fcf2e771 : SUCCESS in 46m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d0563e5279d4f21bffa99bb0d8a66b2 : SUCCESS in 1h 36m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ef986611d4546d2ac9ddb3cd516ce25 : SUCCESS in 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d855217f2304dfe4fae4247354722f6900ac2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:20:37.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a0593a466c0a406caad86686a9ba45c7 : SUCCESS in 22m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68c808f141df488e8ce79f285f890f38 : SUCCESS in 1h 52m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/471edab739c046af8250755012240067 : TIMED_OUT in 2h 03m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/67c07267d265493b8dc21d4d48298f3e : TIMED_OUT in 2h 06m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5946b791ec9149a283cca9609a511f6b : SUCCESS in 1h 21m 59s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/92a72252dbf449df986e119e93e49041 : SUCCESS in 42m 28s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a0ef4939f19148d5a92662d9bdc1ae1b : SUCCESS in 1h 50m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/65c158b729c04fc7a48bed8f70ce6218 : SUCCESS in 1h 39m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ba9bda86dbad466890325c86ac8a191c : TIMED_OUT in 2h 36m 09s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/28fb704dac16463f99b0af25b5528228 : TIMED_OUT in 2h 03m 58s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/36f555c464844fa1a16f11686598d23d : SUCCESS in 1h 40m 21s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/0ced5efcdc884677a97de5529bce6c6a : TIMED_OUT in 2h 02m 20s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/100277d1b8044f7c9751faad230dea51 : TIMED_OUT in 2h 04m 29s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/5323d131c9e84d198b9377e952e8cd70 : FAILURE in 1h 17m 19s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/4d91c1a1a06e408a9447d2babc872007 : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/1c34d50a86f54bf0828450d1086246d9 : TIMED_OUT in 2h 02m 43s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2293ee021d954f39b2bb726c4b863046 : SUCCESS in 33m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d2a9a3048374448eb70c39b4609c3e03 : SUCCESS in 19m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6d88b3436abb448f86dc1f6db38c4093 : FAILURE in 2h 10m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/efa895ab12444b3092c3fa89bd7ec1d5 : SUCCESS in 57m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71ea0da4cfa345f9a6fa9e3734a635da : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ebdfa2b4984317a02b37f93d1c5a53 : FAILURE in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0460808dfe2645689d3c0bb3fc5c64cb : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96d774a490a245538925f62ff2e2fd44 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a063531c16c14422ac46ef8a37da1f4c : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb503f6fcc404350aa8a52a972fbdd48 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec518e3d05f54cf3a4b981df4ce4b74b : SUCCESS in 1h 50m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3fe7bbe0d2f43ef9061499851ae95bd : SUCCESS in 1h 41m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a11db96af96c483881e0683b7321ab31 : TIMED_OUT in 2h 02m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/36410f67e9374f048ac5e9ccf47ac279 : SUCCESS in 57m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6470d58776cd429cbb42056b7389e0c2 : TIMED_OUT in 2h 35m 34s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e49c026253ac454991aa33b23e41ef46 : SUCCESS in 1h 24m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/013a372d99f74c29ab340eb4a43358f3 : POST_FAILURE in 1h 28m 20s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/85b05ec134544fed948c41520c3d5afb : FAILURE in 1h 13m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/43274ce097854636baf5465e6d70e0ae : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/3684bacf0ae946f3a98a9c3a47b70a77 : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/a79c2197de04421ba2f76a91b63ef508 : SUCCESS in 36m 40s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/692b77d0cd184a8893540d0017ca911e : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/29a06290845946ec8bc99e8000b9a94c : SUCCESS in 53m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/4b3ca7014eff4022b033cc78559b40ef : SUCCESS in 1h 14m 58s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/31ab0451854e420c94ff5d9520e61114 : SUCCESS in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b65af7b7912fd27a2e126186213e18126d39fe3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-07 07:36:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"77d2ef7c2194c3eb41cab33046e710ed2ada821e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-07 07:37:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a7360afaede24002746aee714930fe46c93c6305","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-08-07 07:52:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"df82a8a1ba2751d57b23553731b6d2f12aba755b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:25:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d080bfd626f84186a36f7cd7f16af214 : SUCCESS in 58m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4038aec15b0c4e989b57dbc83258895a : SUCCESS in 53m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e37acb4fb2a2466c9f1ed9e7b53fa6b4 : SUCCESS in 1h 41m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b23624ee34742a6a499811ccc464684 : SUCCESS in 55m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f69077d0ff3ce2cccd56b68aa7f4b0a92ee2829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:01:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fb15ddc380184e76a32973a2042fc441 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9a8e3a1e55774b199c4137e3176fcdf0 : SUCCESS in 38m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6fcbd309a2af4ccb801254154bc34de3 : SUCCESS in 1h 22m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5c27b0b205bd47b3a682e8ac75e44a5c : FAILURE in 53m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0352e55360b64884a4d6a676eba3271d : SUCCESS in 1h 19m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d8904136838a496bb931966b671ca7b9 : SUCCESS in 33m 05s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/4bb6ce03c2ad4a1c9fd8ec8f92db2bb6 : SUCCESS in 35m 49s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1ddc5858256b43b7b4ed9cc5a24fef2a : SUCCESS in 2h 38m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6340e33735f54f3cbedf0da90d087f3e : SUCCESS in 1h 59m 22s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c9663a5afd8942c2b3ac0cdd71cf8b23 : FAILURE in 1h 09m 16s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/295a467cb25946918db5c18609a0fe7e : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/545d2b0a71bc407c8b6c65dcfc383d95 : SUCCESS in 1h 27m 24s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b19ddff3c9854d86a7589460bf2a9a4b : SUCCESS in 1h 37m 55s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/5eea7d1386ae45809bc80e12eede1f8b : SUCCESS in 39m 03s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/6552ba66ba474f2c8d45292314533bdc : SUCCESS in 1h 15m 24s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/9bbaef8a0d4d483cbbf9c954823c450f : SUCCESS in 1h 36m 19s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/da90cce956a243989942de1d6ea3a2a3 : SUCCESS in 43m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8ca62e4bed3a412cbc7cc415d6e23fd6 : SUCCESS in 33m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b91872aac13d49309195a23183366801 : SUCCESS in 30m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dab8ccfca94940ad9fa18deaa06a76a5 : SUCCESS in 1h 00m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08584698e20a428ab6f23b0c8d665bd7 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40c3d985ae44117be00ce77f5fa0f19 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/442b9bdd927942aa92848c57c062a49f : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df57108ed4044e638a0646205c20587e : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a847319e97e4cd6a44efc5c1350f516 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa1e0a9628e94127847f2f6c6376021c : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3fba1960f7847fa816cf38a8b24070d : SUCCESS in 23m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/58b2f0a43f9544dc878590f06b855940 : SUCCESS in 22m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bda03591f90c49719d99d3c19311fdb1 : SUCCESS in 1h 01m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bfd0ce797b374856a7d68bcb14bfb7cb : SUCCESS in 1h 31m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9703ce3ff5745df97f4e5711cd3b8d3 : SUCCESS in 1h 21m 08s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/04f224334f1240e38ea858027a10dcfc : SUCCESS in 1h 31m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/15c65452e8364d03836a5be22ae736c3 : SUCCESS in 1h 36m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/dc70e73153df4d54bc135a4a9c677c5d : FAILURE in 58m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c81a22627c9644b7b5907c85355f52db : SUCCESS in 58m 07s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/e97d9dde83be4c978b30918c28778cd7 : SUCCESS in 2h 15m 14s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/c1c00512487f4e74b6ee348ab82bcc82 : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/04f3de04ff1e4ebd89f9342fddf1e278 : SUCCESS in 48m 25s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/fcd34990fa8d41208ad5887eb42b0a7d : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/b23d028303f1465ebb7302ae13d21604 : SUCCESS in 1h 16m 05s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2542d8f3a42446e7ad2c5eae2f20a62d : SUCCESS in 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1644436b356b9f54042df45303c2fb1e27c8701a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-07 17:04:09.000000000","message":"Patch Set 3:\n\noh-ho, looks like our cinder config got broken in the mean time 2020-08-07 08:29:42.771 39 ERROR oslo_service.service etcd3gw.exceptions.Etcd3Exception: Not Found","accounts_in_message":[],"_revision_number":3},{"id":"b257d4b57d8c6fe5c001911ad0bb5506fee09a5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 08:22:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2e10906c4df0d53bbbb4d4ffa3f2a164077fb582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 08:23:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e7a102d560627d26cbcec9c23b35a98d8c99bd3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 09:43:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/49baa8f0342b444893975bf12be92036 : SUCCESS in 49m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7ff875833ea3414eac2f4afcf1590b9f : SUCCESS in 1h 09m 49s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cb7e49688fd84d2c87062315059e57a8 : SUCCESS in 1h 12m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7253eca2b393496b9f067aaa34f56833 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b30951d1cc4f4f8857ceff71b537fb : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04fa0d5beed541c68a9f39e2aa52a441 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55d4b7a9189c49fd8ad7399a24614fc2 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c87462e9fb4624845c834164c13bc0 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1272ac5112d45458f8bc13b0a3e3206 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/87667a5c35e24275945b02ccf2c7d472 : SUCCESS in 19m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6068d9f47cd04db1844f60f954d4dec1 : FAILURE in 19m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/172c290924494600905280b8368d42fb : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":3},{"id":"84d984d98b1dbe3eba674ba8d1ebec7b0c78c345","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 15:01:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"59aa5cd2e5154b7ba5ea6cdf2ec8bb0e9a58f329","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 15:01:27.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9b68f185a3417308ca3f3e964dc773a3f7e0eb9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:48:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1b2136c68624a36bf6bd4887b02eaa6 : SUCCESS in 50m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1a627ae311940d3b15a58e519a9e4c6 : SUCCESS in 43m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58a9859836f3427b9162f32915774e9b : SUCCESS in 1h 40m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bf9927af50d489b9ccc566e5fc5fb5d : SUCCESS in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00f11199aad82a9e6259bfe8e166560dee94e8ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:29:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/06fb840aa553406597e41a9bedadf019 : SUCCESS in 32m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/683ad05c634846e7b5aef776c87dddb0 : SUCCESS in 1h 02m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ba8a4b67c66f4bf286d357630c87ccc2 : SUCCESS in 1h 06m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ddc78d84a1d14113b178f06578ffd015 : SUCCESS in 58m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f83729dd20fa4738a8bc073824e2d30a : SUCCESS in 1h 09m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a7227fe04a0f4e0cb909dd8d6693f7a0 : FAILURE in 49m 31s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/28fa1cb8e3cd4264bc3afabbfc82755b : SUCCESS in 33m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bb672c97bfce48cc9213386539ddc804 : SUCCESS in 1h 47m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fb4054ba06794917b54ba7a33082b5d4 : SUCCESS in 1h 26m 36s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a71bdcc48ff14757bda8b709ad2f30f9 : SUCCESS in 1h 36m 33s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d456981634024643abfc94c134b417a2 : SUCCESS in 1h 37m 10s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c21b2ba46c644243a9edf1743ddd6d4a : SUCCESS in 1h 33m 11s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/1a7f9d02a9ac43edaf3c3b0fd9b00025 : SUCCESS in 1h 31m 31s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/30aad436960344d1a7986837893c5eee : SUCCESS in 32m 59s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/b8064c1ea2f142498eadfc41b6774645 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/3663344a56ae40639367897ae309f25b : SUCCESS in 1h 14m 10s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/cbc8c01470a640b882abae12c78301f2 : SUCCESS in 34m 13s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f1a57b8b1765465cab8d412d5f194c87 : SUCCESS in 18m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f77604a1d2f4431d9e8d1c874c6157d0 : SUCCESS in 28m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/12dd51a6da0b44b389dab24f97a5865e : SUCCESS in 56m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0604de9f4fb84a06b31b28ef2a7702ca : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da792a2eeabc41ddb742885fe1605683 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80185dc20d224d148741e46d51f98e4b : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49802a9af1294c0b8d372d2b8e89adeb : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52427858c74470f8686257124319cfb : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/755909f6c86a420fba5aa907c010a433 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74de591917b949ba8a8ad91e07ffcdeb : SUCCESS in 40m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59ca5d0d42954df49576a0496d0d2317 : SUCCESS in 47m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4333e09063548c0af64fb2a5e6a233e : SUCCESS in 1h 06m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ca329010ba74cc8814519885545222e : FAILURE in 42m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b57023de91e145d09d36530299b19647 : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0c89ab330a004b30b4b1261345b0d01c : SUCCESS in 1h 28m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/49ff85c9a0ee4e769ba67f5550b37307 : POST_FAILURE in 1h 26m 01s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4a37de579577485fa8ba7e55ccaa68ce : SUCCESS in 1h 26m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f3e9d5014f574f24b17b9e6f516023f8 : SUCCESS in 55m 49s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/de2380b80c454581b56833e7b230c481 : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/d9f9a9c2cc5e48a894d7b6510e4f2148 : SUCCESS in 47m 28s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/4d7e8dc78cd44c28bc634e1c4819ce64 : SUCCESS in 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/5f432c5ae1dd46a4b1ec740e549a6ac5 : SUCCESS in 55m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/2074f3e5489d46869d903f99d139b4c9 : SUCCESS in 1h 28m 23s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/51f00c2600db4d47941de647891ab3ac : SUCCESS in 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d60d3a779583b5103186be9ac0f02851cfcfad3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:29:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c17a8c283cdf3a8d20e7e74c5263af3508a4d285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:03:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fa720293ae9f21fa6ac6ddeb3fb86beb1b9c59b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:03:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dd37a4dfc3484245a55abd5370f7e715 : SUCCESS in 26m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/23eb488a05fa4355a90583f18ccc8ac0 : SUCCESS in 1h 06m 48s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/20ddca699a0146db867ca99aedb03ecc : SUCCESS in 57m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82fd3f06b0d04daabfb41b5896a8a3ae : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb399ebac9d84bcebbf3bdd2df19ea22 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f08956f3ac0422888289642861c30de : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9b02666bc44b549c097be71ccedabf : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae1fb688311486687a6ce4a56f2741b : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f2457a63cc84144a9044c2659b4a187 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ad8696acbe84909b029d965c7b90d61 : SUCCESS in 51m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3eb4d214bcb417a8a2a06983c1e8759 : SUCCESS in 1h 18m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64901a8bb99044b8aecb9e3564edceeb : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6357eb3c4cb602627e4586e2ff9632d8335dbfe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:06:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a017f13d095e4e168d1b479afb4633bf : SUCCESS in 1m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ee6b2c70f5a4ed88bd01a71d52ccc9a : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":4},{"id":"17949ec1ecbd9a4eac7b7d3d8fad7749a463819d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:08:49.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit c191e6fbd7ebf0e6f455dc8cfc34ecda64e33bc1","accounts_in_message":[],"_revision_number":4},{"id":"fb95010d14c17acdcaedbe39ad7228aa5600be89","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:08:58.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 797ecd6ddbadcc7a726ce1611c210667565d8a91","accounts_in_message":[],"_revision_number":4},{"id":"3374ccab1bc136c507a8ce1e8b3fe248f8912fd1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:09:05.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c6835b12a6f1e8f715aa5c69383a8d1836551496","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c5320eb22386c34d83435fa0b9aa03b178996b52","revisions":{"eb0b386b8574039c4bfb6b3f5de42f08767fc552":{"kind":"REWORK","_number":1,"created":"2020-08-06 10:02:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/745105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/745105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/745105/1"}}},"commit":{"parents":[{"commit":"53d98f540f3af9972cf3040e529f7d2f14c4bc4b","subject":"Merge \"Add missing horizon plugins for debian based OS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d98f540f3af9972cf3040e529f7d2f14c4bc4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 10:02:16.000000000","tz":60},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are two issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n\nThis affects the fluentd container, which specifies directories as the\nsource, without using a glob.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb0b386b8574039c4bfb6b3f5de42f08767fc552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb0b386b8574039c4bfb6b3f5de42f08767fc552"}]},"branch":"refs/heads/master"},"25ecb6303c123130472ef7acf592924a05bbefa5":{"kind":"REWORK","_number":2,"created":"2020-08-06 12:40:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/745105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/745105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/745105/2"}}},"commit":{"parents":[{"commit":"53d98f540f3af9972cf3040e529f7d2f14c4bc4b","subject":"Merge \"Add missing horizon plugins for debian based OS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d98f540f3af9972cf3040e529f7d2f14c4bc4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 12:39:32.000000000","tz":60},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ecb6303c123130472ef7acf592924a05bbefa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ecb6303c123130472ef7acf592924a05bbefa5"}]},"branch":"refs/heads/master"},"b83a1be7bc3b5ee72f1f199759b53ef1e6686c91":{"kind":"REWORK","_number":3,"created":"2020-08-07 07:36:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/05/745105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/745105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/745105/3"}}},"commit":{"parents":[{"commit":"53d98f540f3af9972cf3040e529f7d2f14c4bc4b","subject":"Merge \"Add missing horizon plugins for debian based OS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d98f540f3af9972cf3040e529f7d2f14c4bc4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-07 07:36:55.000000000","tz":0},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b83a1be7bc3b5ee72f1f199759b53ef1e6686c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b83a1be7bc3b5ee72f1f199759b53ef1e6686c91"}]},"branch":"refs/heads/master"},"c5320eb22386c34d83435fa0b9aa03b178996b52":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-28 15:01:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/05/745105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/745105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/745105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/745105/4"}}},"commit":{"parents":[{"commit":"b4dde581c528c2072e408768601b75a97551e5a8","subject":"Merge \"bifrost: stop sourcing env-vars\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4dde581c528c2072e408768601b75a97551e5a8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-28 15:01:25.000000000","tz":0},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5320eb22386c34d83435fa0b9aa03b178996b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5320eb22386c34d83435fa0b9aa03b178996b52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
