)]}'
{"id":"openstack%2Fkolla~745631","triplet_id":"openstack%2Fkolla~master~I5365c813bab5e3205fd34eb64dc3337f192d67d4","project":"openstack/kolla","branch":"master","topic":"fix-wheels","hashtags":[],"change_id":"I5365c813bab5e3205fd34eb64dc3337f192d67d4","subject":"[CI] Fix wheel url for CentOS","status":"MERGED","created":"2020-08-11 07:37:43.000000000","updated":"2020-08-12 08:29:08.000000000","submitted":"2020-08-12 07:25:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745631-1597217129922-de949bf5","meta_rev_id":"5b71b61742479222f82d109fa9ed409ecd05c086","_number":745631,"virtual_id_number":745631,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-08-11 18:43:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 07:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"recommended":{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},"all":[{"value":1,"date":"2020-08-12 01:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"value":2,"date":"2020-08-12 01:30:11.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-08-11 08:14:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-12 03:18:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-12 05:55:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},"all":[{"value":1,"date":"2020-08-12 01:57:25.000000000","_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 07:44:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-11 07:44:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-12 01:30:11.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2020-08-12 01:57:25.000000000","updated_by":{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},"reviewer":{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},"state":"REVIEWER"},{"updated":"2020-08-12 03:18:50.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-08-12 05:55:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-12 07:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e180eeebd28e7b84a3455dd19df5a408a3ec9a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 07:37:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcf184867fe0f51a010256da7f7b53fa946da2c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 07:37:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b175e5878538c794ba254c7cfca624a69dacfe14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 07:44:42.000000000","message":"Patch Set 2:\n\nwe should really use the parameters per what is in the container, not host, but this fixes one thing now","accounts_in_message":[],"_revision_number":2},{"id":"4af5e93dd575091edd540fa0c49f66ecedbeb4b6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-11 07:52:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\ngood catch, wrong code","accounts_in_message":[],"_revision_number":2},{"id":"1ea15bc705bab6b97467caa17599ba4a7354fe1c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 08:02:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"47bd244f9321020660a1e9908631db8faac84b06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 08:02:18.000000000","message":"Patch Set 3:\n\nah that precedence :-)","accounts_in_message":[],"_revision_number":3},{"id":"acab521ea4ac43d7c182e819d0761ef8340cfce1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-11 08:14:53.000000000","message":"Patch Set 3: Code-Review+2\n\nENV PIP_EXTRA_INDEX_URL https://mirror.gra1.ovh.opendev.org/wheel/centos-8-x86_64","accounts_in_message":[],"_revision_number":3},{"id":"03911bdcb00b49ab329328cd96ac63a0a07fe5be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:58:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/73907c15a12e4e979527c700f462e1e8 : SUCCESS in 35m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1820ffff2094277b66e1309905cfa0f : SUCCESS in 31m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/335f26e261b64c57b14dd952ab5ac166 : FAILURE in 55m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7530681804184645893ebefa5cf5260b : SUCCESS in 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"113218030c50461b46bc81874cb19dd4669adf62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:22:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ab98fece91a4466a9415e3ce84d36a89 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0e1c3b0d0f594cc2b2ee94125ebf5dde : SUCCESS in 25m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/87213a566fd64160985782efc7a79c44 : FAILURE in 44m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3e6d60636afc44e0b1a0c15fede36ae4 : FAILURE in 1h 17m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1753e21a348d45dc91613a663c2e0759 : FAILURE in 1h 18m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/74326c1a1a1a47779512b52e8b5b6150 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc36d934e82d41dfa637be71829325d1 : SUCCESS in 52m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0df84d9fea4c4afe9efac1ad1b301d16 : FAILURE in 37m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28ead34794fe4d1cbdfc22b87e1c9e9a : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8436777ab3a7490c81bee054c22032ec : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fefaa54cf4874b5cb3ac2b09a2e1afe1 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c900647263f42e5b77df9bc0ef61ef9 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759c4113e7094622ac6c7148383ed666 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1ef38eb9d7d45429f5740cd6ed9197e : SUCCESS in 20m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/507390432f504facb2d26b7b9dc55f9c : SUCCESS in 19m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4ef3b5b5f2942c28ce7e66c9379eebe : FAILURE in 39m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49ebd670331149d49f0a24f0e2416baf : FAILURE in 34m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fe1420f75f62460b96a514bec917c81d : FAILURE in 47m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3cfda314922daed400dc01804672636807762d3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 15:34:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f892b0572b736657f90ee009a0145104a5258cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:20:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/17703bb286224986902fc6884665077d : SUCCESS in 54m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24c1e3c3f75244d8b5b7447e8bcb9b82 : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74e80592d5b5438aaa963b0f3a9bf73b : SUCCESS in 1h 39m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ca567bea85648e79e9413145f49eaf0 : SUCCESS in 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70b1a6cc35aa5c0cee11fde825d4378a321f651b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:24:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ca891bc47f61437584d2a2a17493a997 : SUCCESS in 24m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/71501b830404486caa01380f8f01457d : SUCCESS in 37m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c6bcce868aa149b19722a6b80809a467 : SUCCESS in 1h 14m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/987043bdc1c04345a84b4dbc332b5a6c : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ac4d6bab2f464a5d951f45e33ff9b715 : TIMED_OUT in 2h 36m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f9fd88391e304c0cb8f163b66b352462 : SUCCESS in 25m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7d6c01c66644a0f96356148951d8d2d : SUCCESS in 21m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94524021e1a64f3b82d867f54555557a : SUCCESS in 1h 05m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/191b9e70ef5d43799642551e57706c3b : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d556e8c7be4596b1e3ef7a23889731 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1391e969a31a4aa48524fc41ea4f66ad : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526baa24855b4652b718e68db14ad666 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b285eef2e84c4c8b8e8d0d7a16808463 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/394ecfe22f424657aacc569c41ed3fd7 : SUCCESS in 41m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/827cd0ffc1ef4025a3101db2eb35d9e4 : SUCCESS in 22m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/89ee14059b994e529344b7e5e0d2ad37 : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9975ef9d9b546179c074c1eb71b18b3 : SUCCESS in 58m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa073b7ac703405abf569f9c069f90f8 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"d194868a7ac759c496d31bd57b37b9b0d10a839d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-11 18:43:47.000000000","message":"Patch Set 3:\n\nrecheck\n\nunlucky","accounts_in_message":[],"_revision_number":3},{"id":"2ae5b85ffb6029550b20a137613eb13d3959f0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 19:59:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e5748a7ff4214c0ea6d0b82e2c89cd7b : SUCCESS in 27m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4b9401a728ed4f30ae7735930201b06b : SUCCESS in 53m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/de0a721226be443480eb77a19c7f46a1 : SUCCESS in 1h 06m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e58d0123dbed4953beba091540a6283c : SUCCESS in 1h 14m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ec27d0cddaa54d36b5a52f8fa30fe00b : SUCCESS in 1h 09m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/68b615c8be4944f4ab1cc780cd25ffab : SUCCESS in 51m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d7f6b9bc71164f94980f1ce8b599489f : SUCCESS in 16m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a4a0233729e64d70957795ed8e223cfd : SUCCESS in 53m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e56caa5413d7486597bf6dbd243bb8d2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9395a2164f1746d9b40c1a490ae8c384 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b6162365cd4a04926195a66f53a604 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f089b5f7b7749dcb5d535da3fb95d9a : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e044e801384e29844ffcfc9948e326 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65d5f45184c6492191208e446f0169e8 : SUCCESS in 20m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/945600cb4af74778b2ac7d703da4bf07 : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6bde756ba5a49f0b4b397d32b5be8d4 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b56195238aa947de980477e924adce44 : SUCCESS in 50m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07ffb9719afe4cd499e5be44befa3142 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d33dfc982a099ef5a19079db19c1361c48750880","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 20:06:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a87a199e1ae4361b6b2a8f9306bcc78 : SUCCESS in 38m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da402e3d7da24180a1e88faf0a91c40b : SUCCESS in 29m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4416a4fd39243669152232baec82fe5 : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/15c62b4c37514a37a2378ccb2cddf737 : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26df52cb24bc1af14a16a5e1386c0463fc43ba8f","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2020-08-12 01:30:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"232111c7efaa1c2587560bf1193405a2a2296ccf","author":{"_account_id":25905,"name":"Jason Yan","email":"yanyu@yovole.com","username":"jasonyan"},"date":"2020-08-12 01:57:25.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"16436a66bf276f95ecb716b29db43e382f289a47","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-12 03:18:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0b50cced2d08e8074c00a66de9795f91949cf639","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-12 05:55:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"696e56481d8320518811f47d230ae704b3fe197b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 05:55:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7eaccdfa6331209395f5dff18d05928f88b8650e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 07:25:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7b8fbb4cc11423c87c1afa31bf73d3d : SUCCESS in 27m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0e1fe32d0de149d78e90e5ea7158c247 : SUCCESS in 1h 24m 30s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/76daaedc13174304a3a7dd53210ee932 : SUCCESS in 1h 07m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfed3f2065ce446fb9f714489ade5bd1 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0499760f9fc54e44a8fad0e424e45fbd : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c6fa95ad02c4f4f845afff48d05aa57 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f3a2f216b2a4037b99bbabb81e9303b : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2743103e95844d7595c41c9f584360be : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e606153bba344967b75e57a3ea763434 : SUCCESS in 22m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b94e5ffda1ae4ed080181e7f7247f7da : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/59f3db30ea1c4047bbb965b9bf652c3d : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":3},{"id":"2273faf6dee494081655dc6cc75ce36ebad1f7c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 07:25:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"157dc103929fad8364f400d9792bea30becc9ac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 07:27:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b661563dc794ebea38221a3a8331b3a : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3},{"id":"62f4e6ae70ad64cb0c0ae98dcd6b9039f348fd1a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:58:53.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b4faacb4e77f91482d0ea0924574dccc95b399b7","accounts_in_message":[],"_revision_number":3},{"id":"7a4d5dad22fb19b06b9186382c0163f750fc14fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 08:29:08.000000000","message":"Patch Set 3:\n\nfollow-up: https://review.opendev.org/745804","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3ef68d91c60c9d451450ae0e371e25402bb4c6f","revisions":{"1898d39b8118e200fd36913faf13fa0d553756f2":{"kind":"REWORK","_number":1,"created":"2020-08-11 07:37:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/31/745631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/745631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/745631/1"}}},"commit":{"parents":[{"commit":"fd60c657101167b056177ae95aa2c695f14a93d4","subject":"Merge \"Refactor httpd install to base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd60c657101167b056177ae95aa2c695f14a93d4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"subject":"Fix wheel url for CentOS","message":"Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1898d39b8118e200fd36913faf13fa0d553756f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1898d39b8118e200fd36913faf13fa0d553756f2"}]},"branch":"refs/heads/master"},"65b66fbeefd65322347d21c178bf662ffdd22e55":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-11 07:37:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/31/745631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/745631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/745631/2"}}},"commit":{"parents":[{"commit":"fd60c657101167b056177ae95aa2c695f14a93d4","subject":"Merge \"Refactor httpd install to base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd60c657101167b056177ae95aa2c695f14a93d4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:37:44.000000000","tz":120},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65b66fbeefd65322347d21c178bf662ffdd22e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65b66fbeefd65322347d21c178bf662ffdd22e55"}]},"branch":"refs/heads/master"},"a3ef68d91c60c9d451450ae0e371e25402bb4c6f":{"kind":"REWORK","_number":3,"created":"2020-08-11 08:02:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/31/745631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/745631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/745631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/745631/3"}}},"commit":{"parents":[{"commit":"fd60c657101167b056177ae95aa2c695f14a93d4","subject":"Merge \"Refactor httpd install to base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd60c657101167b056177ae95aa2c695f14a93d4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 08:02:06.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3ef68d91c60c9d451450ae0e371e25402bb4c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3ef68d91c60c9d451450ae0e371e25402bb4c6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
