)]}'
{"id":"openstack%2Fkolla~745801","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I5365c813bab5e3205fd34eb64dc3337f192d67d4","project":"openstack/kolla","branch":"stable/ussuri","topic":"fix-wheels-stable/ussuri","hashtags":[],"change_id":"I5365c813bab5e3205fd34eb64dc3337f192d67d4","subject":"[CI] Fix wheel url for CentOS","status":"MERGED","created":"2020-08-12 07:58:53.000000000","updated":"2020-08-13 15:08:58.000000000","submitted":"2020-08-13 15:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745801-1597331238379-b7eb80e6","meta_rev_id":"2bec153573c2fbb0c38936a0d96990b4e89a6e30","_number":745801,"virtual_id_number":745801,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 15:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-08-13 13:30:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-08-12 09:28:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-08-13 13:30:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 09:28:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-13 13:30:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-13 15:07:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed0cc7a61fbec90b420f4dfc8448ea0d0e34bef7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:58:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"382945a3402df486d5b244b247e37f9e48cffd4b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:58:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9d6470a446e5767d715cf6d40f3367d3b89677e3","accounts_in_message":[],"_revision_number":1},{"id":"0402175a0ce9bfc51f228893224bdc36a2fa7ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 09:24:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7c6dbeee3b7a4df1a0a32efff28eeb38 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/289d6c4f2ee6440d8e3f3f2807aac88b : SUCCESS in 32m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/efe608ceafd140feb61900d5ac6eb0f1 : SUCCESS in 1h 14m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/03217b52376b44aaba612b4d85547c4d : SUCCESS in 1h 03m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bf530cae3af74f65b9cf274d3f0955e8 : SUCCESS in 1h 20m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a90cbee279f3408093907f7b652faf81 : SUCCESS in 23m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/90418d39d1e0487082021c9a39f54fbd : SUCCESS in 18m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6d2145cd6d44d9599d206033384ca2b : SUCCESS in 58m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc275c532ae4d3fb29237287d235274 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79af8da57ebf40649f801974f6aa1da2 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d62d7361e9419aa9f5b138df54f14c : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc8b05aaf3664613a37ed7e79cc51d5d : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdc3bba4abe4baf9baee0106e413b69 : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/257dd8ac66a74394af5e9cff1bebcd2e : SUCCESS in 22m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/065cf499ec1740a1bdc7dd8fcd223c5f : SUCCESS in 28m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5fd88012a57a45e6aabbcc30098fc738 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ca6a52f6f91441eb3643af7fbfbacc5 : SUCCESS in 57m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ad0d5890e824bd1aff4a30077ff9592 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d391e6902d1d18569d8a899dd78347d02b7740d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-12 09:28:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f4997ce0378a5d517ed5fdb9ad2fce7ec8410ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 09:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/75bf6df70c774bb4a557dda4b4dc8e9d : SUCCESS in 42m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5fa84858ab6d452aae0445296fad834d : SUCCESS in 36m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a842e02fc4d9453b867e809d97dd1a0b : SUCCESS in 1h 33m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ab70a94874242fa9a9778b87df34fa3 : SUCCESS in 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6112ee1fbb4c75ea981e41734e20d0f09dd76c35","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-13 13:30:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2edaab8dc48184f5657139d45dc4807eaa3bbab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:31:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca9602426d17022c719e90e974e1d03f33aee556","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:07:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f843683a8eb299d75581fc4d43cb1688673712a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:07:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3c00a92e8b1d483fb5fea2fa50703db2 : SUCCESS in 28m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/903614c7dee74aa992be242a97edc550 : SUCCESS in 1h 09m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2304dc80ddf04fd6baa9b641d4bbd6dd : SUCCESS in 1h 30m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18f38b3cedbb4216aa1cecb09c99dc1b : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1f27a08f234e0ca0a31999f344adee : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93c301d00a3d4595b93d594c66352e85 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae5a95f123464c57b0b3f647f19dfb88 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193484a296e5443a96df93160ec63a31 : SUCCESS in 15m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9316cc5a65c9455e94fb4c605adf0b2b : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/35faea08a81f4b52948dc167ea1b4d6d : SUCCESS in 1h 07m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96aba0522b1c440bbacfeaec2b4ff6ed : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"113075ef6bf152093d7b67fe97c48ada92f19c95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:08:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cd1c48eb18439c8015e9de25b4a0f8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4faacb4e77f91482d0ea0924574dccc95b399b7","revisions":{"b4faacb4e77f91482d0ea0924574dccc95b399b7":{"kind":"REWORK","_number":1,"created":"2020-08-12 07:58:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/01/745801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/745801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/745801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/745801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/745801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/745801/1"}}},"commit":{"parents":[{"commit":"c305746b65a98ba9f70b9b20c37e33a4ac6edbfe","subject":"Install Monasca plugin for Grafana by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c305746b65a98ba9f70b9b20c37e33a4ac6edbfe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-12 07:58:53.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n(cherry picked from commit a3ef68d91c60c9d451450ae0e371e25402bb4c6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4faacb4e77f91482d0ea0924574dccc95b399b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4faacb4e77f91482d0ea0924574dccc95b399b7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
