)]}'
{"id":"openstack%2Fkolla~745802","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I5365c813bab5e3205fd34eb64dc3337f192d67d4","project":"openstack/kolla","branch":"stable/train","topic":"fix-wheels-stable/ussuri-stable/train","hashtags":[],"change_id":"I5365c813bab5e3205fd34eb64dc3337f192d67d4","subject":"[CI] Fix wheel url for CentOS","status":"MERGED","created":"2020-08-12 07:58:58.000000000","updated":"2020-09-03 22:15:20.000000000","submitted":"2020-09-03 22:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745802-1599171192964-038e877c","meta_rev_id":"076d7fe29f2b4387ab65baebdeab0a175cf0c821","_number":745802,"virtual_id_number":745802,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 22:13:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-09-03 18:12:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-03 18:12:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-09-03 18:12:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 15:18:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-03 22:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2733d4afd98332bf602714b10b6393cefbfc6c83","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:58:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"454d7755c40e8cd2695ac4282c19382e6f68ee9c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:59:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 25f72694e357c662ce075e11d536b2b4a259291d","accounts_in_message":[],"_revision_number":1},{"id":"76941dd3eab7a18a173e7e81952a25807244ee67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 09:21:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d5b84502fede4f4c94df8fd3811bca35 : SUCCESS in 36m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6e1bcea5054c4b208408213fa2a8effc : FAILURE in 52m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/783198bb48164c0280479cb967322703 : SUCCESS in 23m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d46ecfa7c94342ffaa52732799b65d63 : SUCCESS in 31m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cdeb49e0b4f94440aebf3f60a8372444 : SUCCESS in 1h 09m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bdad1267742f47f6a91cf115f9457b37 : FAILURE in 41m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7eb0576b98ac407b92dee36bcb431349 : SUCCESS in 1h 09m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/14932b2fb3b24b9999d73db20942ba5a : FAILURE in 41m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e39be14544db49698f661dd0ec70bd7f : SUCCESS in 1h 20m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4807c06411f040ff83c84193c3690084 : SUCCESS in 20m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/90cb9d3fd3bb447b958b09f1e79e270d : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e21e6103fb34610b8e5384957ad5dab : SUCCESS in 54m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e27623f4fc0f47f2b0775c3172e8c140 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8804df297e54b6b9b44bddb61dfda75 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4f88c91c35f4a118ea056841ea4be4e : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6ff35c722a469799d85b77f04c44ca : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13981ffdb5f241c79f2c889727681e09 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ae70e8cdaf4f0aba54346ec6608a61 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ff33ad0e9f54e5fa985ffa649a914a1 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/005a321177a049ad9f3acc7f0bba66ea : SUCCESS in 22m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16e67261301846cd9e72b9a4b6cc1641 : FAILURE in 36m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25a3bb9482834d76b3cc4a6593e3f4d7 : FAILURE in 37m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4a7c449aa8604dcfa3f43899f6b5d8d7 : FAILURE in 50m 55s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"80417822f88f96f35b471bd1a361496cb6110419","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-03 15:18:49.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7fe0c02fc23bba2f7fb71e2400f4ec776957543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:57:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/50b51c86be014170a840904c3a4cd4aa : SUCCESS in 35m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3ceaa69eac724b4b9cd27e64c5329d3b : FAILURE in 59m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/33af80d2d6124bd8a80d3777cfe856f0 : SUCCESS in 21m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9de88637d3a94ddf801a12d02fc35229 : SUCCESS in 29m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ae357914bfb84f3db84861be38028407 : SUCCESS in 1h 17m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bf3223bf3a1a4bdfa4cdfc020382aff3 : SUCCESS in 50m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3b2e9e6917784133b40cf50fe9301b7b : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b0841ad838db48729fe6d69569976773 : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36f6733739004ef3b4599b5ef1e00a33 : SUCCESS in 1h 29m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60c2ae0ede3740d68192b93e243ae592 : SUCCESS in 23m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ea71c71c640c4ad6a3eba662a5bc4522 : SUCCESS in 20m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/69510f6ca52c4d23ae3402ee3aac3a3e : SUCCESS in 1h 00m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea4874ae137477fa015dfdff430578d : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbc954782c54e1fadcbe68585907eff : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3509ed3c79845b0965686b6976cc7e4 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78e241a0b5b343ed8bf330d7764ef628 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/076f046a34864eb5a8711d515e47edd7 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ece36dc353c4e93864749163f21a82c : SUCCESS in 8m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c7ab14d717948edbff42226385461c9 : SUCCESS in 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cd0ba094143439e8c0e22873e47033a : SUCCESS in 24m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/40120dbdba034397b23b7cdb9473f760 : SUCCESS in 46m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5aba93aa491427e94098f8d5a5dce77 : SUCCESS in 46m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dc9f3b11e824fdf989a8b1c03d7baea : SUCCESS in 1h 16m 22s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"c7af49a00b0f8d7a030724f01072bf6f866ae30a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e974328e2a6417977f541b1c5d6852a278904792","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b27cf5fe8b57dfc2ee944ca50caae4f209f23ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:48:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d840d33fa7e84b768e540e05ed65124f : SUCCESS in 42m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/149ae9dfa4fe470798bbb4a902b12b41 : SUCCESS in 2h 28m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bcaaa46752a94e4aa7487f4de238bcfc : SUCCESS in 39m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b3bce850eed24e2496b60e354ca89fe3 : SUCCESS in 26m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f9f169edc6b743f68e65188003baaf6e : SUCCESS in 1h 24m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bac5f1a1fa07433d8a4ca7000ddc6655 : SUCCESS in 1h 08m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/148d16e6414f4e6db3f9aba3bcc59fe9 : SUCCESS in 1h 05m 21s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/05b53b3ad66a4919b2e8eef1ba63f07e : SUCCESS in 1h 01m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36487519e6e2467b8d3b51b36473903b : SUCCESS in 1h 54m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/93b55e3937d647f183b707e2398e486d : SUCCESS in 18m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/70dc0b0be73743c199d7a17b095ff088 : SUCCESS in 18m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9e6e789e94a4a70b5f03afa8f547624 : SUCCESS in 39m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23795b76462f4f1c9a05ae4c00f7783b : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036923b64f884a9b9a0ec03bd36c13dd : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5811caceed040a7a14508fc0dfeeac0 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f0fda5009574c9685fdfefc50d13f17 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b425e493cf0248439fd018c5749428ae : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c04a8177fec4ec98a6eaa6522c5a6c6 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7af2fe7f3d88480db75c94842cc5e818 : SUCCESS in 45m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f78bb45b48848a1b5bae8abda64dd13 : SUCCESS in 1h 06m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2129f9bca4e34b13931a063a42ed7103 : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de19dc02798e463db544b97a93b1ed3d : SUCCESS in 47m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b62e21e5eae4c488990622cdf16e239 : SUCCESS in 44m 16s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"e4652d69303dd48dcffa9861db020414005ed7f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:54:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5475317f642bea4db585b3183edbf0a7c14d288","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:13:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bcb437936bd31ee4a0bf76c3dcb431372e7f8ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:13:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7b061278d58c4be88dd82b797744c2e9 : SUCCESS in 33m 21s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5be260e32a7d4283a858e56a5d446e8a : SUCCESS in 28m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4145a14eceb44e02937b6d2d31a16fae : SUCCESS in 49m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a12e2b94a1d4303ad6436cf53a80395 : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fcd84081479a405d90d0ccaf1f36cac6 : SUCCESS in 54m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35f54579f1204160b50b39d6bda37f3d : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38362ac2e7d4a9aa317785e1e859f9a : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f396f8a3aa6a47abb08cea3fcc4c7fec : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9043fb53c5a14f309c3e49e0b19012c4 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef100a7d3d5b4c008d0faf83af9fa407 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc09cb130a447d6b1d47f3c0a94a1bd : SUCCESS in 7m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9ad6e0054fa4c91bd4b6e757cc46d78 : SUCCESS in 34m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1e0b73f323b40c2840f9956074cc3a4 : SUCCESS in 58m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/057b82a427574d19837c5e3a337db651 : SUCCESS in 1h 10m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"13c5beabf3f6b31b357b4aa5b7021d59b4d1cb1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:15:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273042e917a949dcb38f9feb5e57572e : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a31836265efc6d42640dd7bc97bce0113441fb5c","revisions":{"9d6470a446e5767d715cf6d40f3367d3b89677e3":{"kind":"REWORK","_number":1,"created":"2020-08-12 07:58:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/745802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/745802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/745802/1"}}},"commit":{"parents":[{"commit":"e3e2da2e92302a7ba2eca2bc394529b0eeb24358","subject":"Fix masakari-monitors image on CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e2da2e92302a7ba2eca2bc394529b0eeb24358"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-12 07:58:58.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n(cherry picked from commit a3ef68d91c60c9d451450ae0e371e25402bb4c6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6470a446e5767d715cf6d40f3367d3b89677e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6470a446e5767d715cf6d40f3367d3b89677e3"}]},"branch":"refs/heads/stable/train"},"a31836265efc6d42640dd7bc97bce0113441fb5c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 18:12:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/745802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/745802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/745802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/745802/2"}}},"commit":{"parents":[{"commit":"326823c4d3661ee5cc329857eab65148aa1d8fba","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/326823c4d3661ee5cc329857eab65148aa1d8fba"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 18:12:13.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n(cherry picked from commit a3ef68d91c60c9d451450ae0e371e25402bb4c6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a31836265efc6d42640dd7bc97bce0113441fb5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a31836265efc6d42640dd7bc97bce0113441fb5c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
