)]}'
{"id":"openstack%2Fkolla~745803","triplet_id":"openstack%2Fkolla~stable%2Fstein~I5365c813bab5e3205fd34eb64dc3337f192d67d4","project":"openstack/kolla","branch":"stable/stein","topic":"fix-wheels-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"I5365c813bab5e3205fd34eb64dc3337f192d67d4","subject":"[CI] Fix wheel url for CentOS","status":"MERGED","created":"2020-08-12 07:59:07.000000000","updated":"2020-09-03 17:18:54.000000000","submitted":"2020-09-03 17:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745803-1599153401468-33708db1","meta_rev_id":"5a69585be1efedb988aadbaf062226a79ecb9bcd","_number":745803,"virtual_id_number":745803,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 17:16:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-09-03 15:18:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-03 16:19:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-03 16:19:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-09-03 14:38:47.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-09-03 14:38:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 14:38:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-03 14:38:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-03 15:18:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-03 17:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05d4be25f8f62809810b63e3f2da8dd6d350b1a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-12 07:59:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"933db8ec911253e6ad15e29837b5254de35c3baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 09:32:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/11f5565ec4144516bac9e000091b8c7a : SUCCESS in 44m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b763a51a9f244a0e8bfeec924e57a00c : FAILURE in 56m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a3decc2c78cc4afc93ccf2561e7b49da : SUCCESS in 1h 25m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d36cd7a73cd343a99e198bd7e06727ae : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cfa0b83734c94b33992b7223732703a9 : SUCCESS in 1h 28m 46s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/840330de0c4047778529aea5b13b1760 : SUCCESS in 42m 48s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8a8de8f15f41486a926ae2c1ea7e0bcb : FAILURE in 55m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f94964a867c746f4922f0e37cb410429 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf226ea47d44092a741daff71e1014a : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/237681faf0224521ac88dbcac2a300dc : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4371774c1d94460f826be666d17e9f31 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f6a6d8349bc468794f4d5353f801c37 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87e3582913a44d3a83134781a46b959 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73d867f240d544018be88c3100e1c04a : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7a1d60a01fd4d49952bc47161dd4331 : FAILURE in 28m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69bf17f084264ce69ccd4cc2dd36740a : FAILURE in 45m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5178ad01efd1408f8b38544295016c17 : FAILURE in 31m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d6af77938df44cda7fed6cb7b8ae8cd : FAILURE in 43m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f1b6b35d4346db0110b23fdb550e6a91b6f9fb6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 14:38:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"65a180565603803520cd729851a2202a4f057a2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-03 15:18:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5b3a11a5a13ae43d68c05db5ed00317edaae8977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:08:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/17b7e5131351426887e06b51382c1460 : SUCCESS in 37m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fc4c786218c24d5ebc1cb5fc8113ea05 : SUCCESS in 53m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/586573e8b35c4b408a08a4ee7a77ee48 : SUCCESS in 1h 24m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b8753f9317b84a519c41ba69c1bc3791 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f3be4f45fe644d9abff07baa79f9f8ee : SUCCESS in 1h 27m 33s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/383d86f5aa7640509e4e10f23e8968f0 : SUCCESS in 1h 05m 19s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/25d2aaa21ad14541b49e6bb33bf9cfde : SUCCESS in 52m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c39f483ae4684807be694f3d98f8d7aa : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f128f627434542e3995009e1597765cb : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6642b3ef851948cc8c5d0324782aac7b : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50de0b5c31a94367b76ea0aa423568e9 : SUCCESS in 10m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50cb417238614e30ae5ff3af6c924528 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19191c538064cb7890623372cca5765 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a3e9424d980e4cf391018a95b9f91aeb : SUCCESS in 23m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a955faf215d4cb6a7e134db31710308 : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b296f86f2a604ae98c49e61d1be5e712 : SUCCESS in 1h 04m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57ba2538dddf4ec3a76580b0e4b56da4 : FAILURE in 30m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b1ecaed5d5cf4de98e4aa1f1e3ced05a : SUCCESS in 55m 26s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"95d31e869f792b3dd0f040a9e1fba02a10719852","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 16:19:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47a9dff0d6cc172d469d48ee59f9b0afc1cc5326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:20:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b4a9caf697f7d57d530004399f5c2f7ca12d50d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:16:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6a72a063912d8a60cbaa84f270f3462f3e127994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:16:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/42ddecda0d1a480693eae681e5890fb6 : SUCCESS in 36m 57s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2ad90ced848f4b41a87712cada1f4984 : SUCCESS in 41m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482e97ac18f7495b9936d6167ab82ade : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f910c84cbc4d80a12322327ee72a45 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/866eb7ceae664f1d85d78e2fe8fde9b8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c5e2e1e30624abd8d36125450a7e8c0 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a2cb4b83526423ba2ae9b539bfa4105 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2263060f9b4492e819ccc63e623b531 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/24a7f2601fbb40f4a7ece45d4396e26f : SUCCESS in 21m 19s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"6caa63313648c41946760abbe00dc28fc69c629f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:18:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a0ca5bc9034dad94256e185e8f377d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8123bf6f385061ea7180efcf0d359adcea7a57e","revisions":{"25f72694e357c662ce075e11d536b2b4a259291d":{"kind":"REWORK","_number":1,"created":"2020-08-12 07:59:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/745803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/745803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/745803/1"}}},"commit":{"parents":[{"commit":"7038e3ff03c6d08e7b794f9087675964ddbbed0a","subject":"Drop systemd support from nsswitch.conf on RHEL-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7038e3ff03c6d08e7b794f9087675964ddbbed0a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-12 07:59:07.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n(cherry picked from commit a3ef68d91c60c9d451450ae0e371e25402bb4c6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f72694e357c662ce075e11d536b2b4a259291d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f72694e357c662ce075e11d536b2b4a259291d"}]},"branch":"refs/heads/stable/stein"},"c8123bf6f385061ea7180efcf0d359adcea7a57e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 14:38:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/745803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/745803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/745803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/745803/2"}}},"commit":{"parents":[{"commit":"3c435cb0781b343ad7f7223609cf2eefb2f8738a","subject":"Fix ironic build issues - LANG related","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c435cb0781b343ad7f7223609cf2eefb2f8738a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-11 07:36:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 14:38:11.000000000","tz":0},"subject":"[CI] Fix wheel url for CentOS","message":"[CI] Fix wheel url for CentOS\n\nCentOS wheels do not differentiate between minor versions atm.\n\nChange-Id: I5365c813bab5e3205fd34eb64dc3337f192d67d4\n(cherry picked from commit a3ef68d91c60c9d451450ae0e371e25402bb4c6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8123bf6f385061ea7180efcf0d359adcea7a57e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8123bf6f385061ea7180efcf0d359adcea7a57e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
