)]}'
{"id":"openstack%2Fkolla~746612","triplet_id":"openstack%2Fkolla~master~Icf7e7bfddec89c4aec89410884480539b6bbfd74","project":"openstack/kolla","branch":"master","topic":"ironic-pxe-multi-arch","hashtags":[],"change_id":"Icf7e7bfddec89c4aec89410884480539b6bbfd74","subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","status":"MERGED","created":"2020-08-17 22:29:48.000000000","updated":"2020-09-17 09:19:33.000000000","submitted":"2020-09-17 09:16:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746612-1600334181913-193fd9bf","meta_rev_id":"78e4dfb9c4f1f0bc2badaab57cf1fa531c6e13b2","_number":746612,"virtual_id_number":746612,"owner":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 09:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2020-09-06 09:05:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},{"value":1,"date":"2020-09-06 09:05:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2020-09-17 08:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-06 09:05:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-15 07:52:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-09-17 08:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 13:24:39.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-08-18 17:44:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-28 05:58:05.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-08-31 09:39:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-03 15:07:30.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-09-15 07:52:08.000000000","updated_by":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"reviewer":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"state":"REVIEWER"},{"updated":"2020-09-17 08:03:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-17 09:16:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ce3dfa0d5ee28b87de40eb82febc0a5351f2598","author":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"date":"2020-08-17 22:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27423c786c220d85f1034ef6b10012bb4da38c67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 00:11:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1b503a5f7c842729ede8d20817802fa : SUCCESS in 49m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77817f7142ff4eb1a63cf87f8219b758 : SUCCESS in 47m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd9b0e5a137b4ed599d48ae2ec75461a : SUCCESS in 1h 34m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8853bded1d2442f99735d7105c567169 : SUCCESS in 47m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a39c36781d3797f8c0aa050bfb48a3f657ea2962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 00:34:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8e7e8800b0d14b6084d5a8524bd93fb8 : SUCCESS in 1h 04m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2e3df3ddd6e74acca50b062e8656cb31 : SUCCESS in 1h 39m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b52dae8d9a26453ea597a39a6df6bca7 : SUCCESS in 1h 37m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/580fa18c0b7a4c799329df65f7b120be : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d7f12e3c1c0d47fa849e1716cfc27ab4 : SUCCESS in 55m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1f9fea3359164afca464c9d8a68778a9 : FAILURE in 1h 35m 00s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a7dd629e5de94622a008597ddfd6ad24 : SUCCESS in 2h 02m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a1ada8d84714db1b97b7aedf08136c7 : SUCCESS in 19m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b8475ecd0d24b93b3a47b08bcce7c0f : SUCCESS in 45m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a39fbe2cf2bb473cb460097992fb8007 : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac10c35c396d4baa982f56ff7798fc3a : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b450b13994084f0fb061436c8922dbf1 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041d11d3444b4dbfa2b62cb7ef19a563 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b1c47a70c794692b3b0b952eff4aeb0 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cffbda1198548fea1e2134cb4060046 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0bd415d514b44d2e82960729361e7b50 : SUCCESS in 23m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e27eefeebcd94545a0e4468479c86bd3 : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1714f25c5727402a908e9a7b21444b92 : SUCCESS in 41m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b88102acac74c77a821c08a01a70f81 : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ce42face44514d2fbb7c30912b13b6b7 : SUCCESS in 1h 02m 09s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c04a9f43d36541e5899e17ecc0d63951 : SUCCESS in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c59fcc936fcdfa07d4d2824008919e3f6bf93d4a","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-08-18 13:24:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4839b6749b59b21e79073d1a00355ff5e7ed4023","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-18 17:44:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3379c8a53e7e16e4e1778195d1e9fec2a29c46e2","author":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"date":"2020-08-19 08:25:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b012897d19081be75942890290bb9c223c5e6f8d","author":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"date":"2020-08-20 17:40:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"276b9e94925a3490087919cb5ee5fd2d052bccf4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-24 07:28:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4499b463dac4f6485abe8ce4f4a247dd1d863d71","author":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"date":"2020-08-24 13:29:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cba72a69d04052058ee5b03a44a861fccec786d","author":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"date":"2020-08-24 13:29:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"311019c3f82c62c7ea6e937090dadddc87755386","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:58:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9836115334054deabd0b45797e785628 : SUCCESS in 40m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a2dccaebc5d644e389648120fbc3a8e8 : SUCCESS in 29m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/babb54b408904c21aaf04eb0a732ef6a : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/abf7183bae7a415a8a70d49afc91b608 : SUCCESS in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64bc149b9d2760e65b8779fa682093aaf2ad6923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:01:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/015983e8a5e74b10be726dfe728464fb : SUCCESS in 24m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/222a4aeebbfa4e20943b30a78be67735 : SUCCESS in 24m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a694cbf7227641b9b6fff07d638960f1 : SUCCESS in 59m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7853d7ec6f63477f9e9a0313683298fc : SUCCESS in 1h 00m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/da0c73d1e5fb4004b25b876bfd7531c3 : SUCCESS in 1h 26m 13s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9cfef3d0458d42ad835c6966d6746497 : FAILURE in 40m 21s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/eb60ffb30f06483ab8921983f4fe8fa5 : SUCCESS in 53m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a2f2c5888f742bf91370cfe80539d15 : SUCCESS in 17m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aedc84307207493ab015982682908dca : SUCCESS in 24m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b78693ee01c846a6a33110c4ee4a4f6d : SUCCESS in 1h 02m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a0e229bae7d4cd687c0e8ff13700694 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ab55a25e1c4432955c0ce8aa8522b5 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed687ff0b9a14ea0867a2e10522a1bef : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae220e1391f445bdb6577a244f6aa83b : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f991f615cc564141b03fea770f9cae1e : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7bfeebbc12944d48a352d28d8d41cef2 : SUCCESS in 17m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf96c0fbd86b4eff8970b8c560ec63ff : SUCCESS in 19m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8175ca3dba17464384383aefd472f42a : SUCCESS in 50m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f56b96975e034cd09494fe207142a5b1 : SUCCESS in 39m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d50a63c90b14da0b8244cea3b3553bf : SUCCESS in 1h 06m 50s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1234df34f0464e5d97d0a296ee5fe8b1 : SUCCESS in 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d642521fc38ee2a4e65d1d80cde0c685a37c639","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-25 08:26:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5090900753562246099e24f9ad8ad87d1ae1b6fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-25 08:26:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"31f167b9272f598ff2026162dfc558730e167f7d","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-27 15:53:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ed2186c9d05e0e46d3057e25f3e98d0e057160fa","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-28 05:58:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"826ddd081970ad1e8acdeefede383b4b4d20afb4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 08:14:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nMissing release note.","accounts_in_message":[],"_revision_number":2},{"id":"e2921c268e780f7d6ac3980ade8e33acf14bc8e4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-31 08:57:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd2f698b660d73bdfdfb962458ba71e74fd439ae","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-31 09:36:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\n+3 (see the comment inline)","accounts_in_message":[],"_revision_number":3},{"id":"54bcddf440a6569aa98ef4e51e8225c0b1fcf61c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-31 09:38:18.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6f8b5329fda53d9da8fc5abbfabf15f85a72416a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-31 09:39:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cff3e51bbed022fafd21019fc8c5823fc4180e73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 10:58:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4623c9f630f4909bd592933061b50bd : FAILURE in 35m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/116b9355d9b04944bad4973329dfdf66 : SUCCESS in 28m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04bc8ea97c4a4891807aff878524681e : SUCCESS in 1h 15m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/56b00c71027648f08fde8b01ceb90f5b : SUCCESS in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9cfc38756205cc20d5d3894458b93b5289a25199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 11:04:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/36105c18e52a4ab88e77d3419e99ed2d : SUCCESS in 17m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/650351a8deb9471083205d5254ff8619 : SUCCESS in 19m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/27b63df8d7e749a58a54994b5cb76b52 : SUCCESS in 49m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9ffc3b85c1ca4364ae3125033a2cab90 : SUCCESS in 47m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/51c22b5097544cceb09b4037c7f64be0 : FAILURE in 29m 17s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2a2c49d815394bf6ac68c9e30a789951 : FAILURE in 1h 24m 33s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/061c3457e1b9425e861a3022677ecc4a : SUCCESS in 49m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2c0d00425eab4cc08d82dc9241c25e97 : SUCCESS in 41m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9710c47c35bc4c7393fbba77dd0a2d1c : SUCCESS in 13m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5c4e3787e3c44e3cbb422c3f9955ff20 : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40f80e5f5e6c419a82ae36805409f1d1 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9bcda0b8814d61a3495684b4e4c110 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2154070bf43f471e90d38fde248568b6 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d3c3442af494433bce68008281ebeda : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ceb8a9cb3d4049bc262144b0b405b0 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1a18be8311941b0ae6dcd8b08d70af7 : SUCCESS in 16m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ef60ce551c8492287a14b56dfbedd03 : SUCCESS in 16m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07f83a99ea844ff28bf25fc52b23f2d6 : SUCCESS in 41m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2504b8271430421798fbbb356f486241 : FAILURE in 29m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01f0aaf2acd14a229b53d9718926a48a : SUCCESS in 46m 25s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0a49957fdd984cc0bba9d23aae49d2a7 : SUCCESS in 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53102834301a327d44a7476913236aaaf10ef320","author":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"date":"2020-08-31 13:12:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a7c6fe800cd98c7546864329070181a6ff2f8ca8","author":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"date":"2020-08-31 21:12:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"db176203392893f9a720c64f051b692283ca29b4","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-03 15:07:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"824c2ec40835d8999283e200c18103f258971692","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-06 09:05:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5a49dd65241cd9c47e7fa2b93a6edc92fa35d4e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 11:07:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b716786f1c224001aaa3553df919eb3b : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ef60078fd124e0da55d8d3983c3c659 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3751219011704710a09cb9b9a4fa4bbe : SUCCESS in 1h 52m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38cd6b72526b419a8801ec69d46206ae : SUCCESS in 1h 07m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56a83d6e6c78b177f9e16c9113a37f48504a88a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 11:13:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7df7af26dc3841e9ae95897e65648068 : SUCCESS in 49m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/46eb82fb4f064fa89cd2f34a5c07e9bb : SUCCESS in 21m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3719e168ce894765a656d166ca349524 : SUCCESS in 1h 11m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/33b490c4f9b64f2db146d92d30e394e7 : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6c07856cc5f54e0f89ff0c8de487bd89 : SUCCESS in 1h 03m 16s\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/aa77d17dde1b4e8f9acdd9cee72f7164 : TIMED_OUT in 2h 03m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe02911223c44a7a865f9b5ff4410157 : SUCCESS in 21m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/26afdf6831d44fbb9599a469cb57d3ce : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/237cc76b66ee4055a2161a3b454f7a78 : TIMED_OUT in 2h 04m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b76c3a40f2403da9728faa2217c586 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28754d58893e433e9b3c1200c17754bd : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb6ac32f4940459c98abaf3e8cb9af05 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcde741acd654724972ee49855fe8f86 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288de63bd57840ab938cdc3868a4ccee : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ca514a25c1a42d48f883f469f3b54c8 : SUCCESS in 18m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ecb5a8f33bd4c3a930daf3d1e442c6e : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9870bf3e8fca4e9eaccf72d097d9e8c2 : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d9a6f602ef5494e9f39c4073c0a705c : FAILURE in 43m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0275ac1fff354af9bc97bf04decaff86 : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c7fd8fe1296249d78f80620de24ec39e : SUCCESS in 54m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9e6072cfcde722038d8cef89a110c6ca478161ab","author":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"date":"2020-09-15 07:52:08.000000000","message":"Patch Set 5: -Code-Review\n\nIs there another +2 and a +W out there for this patch?","accounts_in_message":[],"_revision_number":5},{"id":"8cfe32ad62fe701167ec13810103f66da0c1fe3a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-17 08:03:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d65d451efb9fa92ebe820930925697a3007aea00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 08:03:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5c9f223ce714a38cb7330232e9cffcf0068c526b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:16:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cd20cbd0db272f28bbeb0660387fb124f587fb14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:16:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7e32201171ad45a78b7efa80a9757d30 : SUCCESS in 26m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/46015901e9554847ac0142c1b4e09fcc : SUCCESS in 1h 00m 49s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7edca0aba1d74adcbc4c93c5efa31834 : SUCCESS in 1h 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b47552a86b71488cb7c48cdf5e38ced7 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346483ec3af748269160e775fd5d4812 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cccdec1753e4206973ce3b566ac094b : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ccc1ddb64a844c58bf0b6e9127d4fa9 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a32fd78a974baf9b30e4fc64fbac80 : SUCCESS in 15m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/512fc07254734957a707800ff88df389 : SUCCESS in 20m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/621008891cf04b479a022361cdcfe3a0 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e723c1aa625b42f3a178f770665078e3 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"3261030dd54f41e607589c04e29939bd371ae3bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:19:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091bd67c8a5448ed8430135cd53b1d33 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c3a3c7bee0d5fd68ceef1446a43f19b2d4512614","revisions":{"38e2421a1de1259d751d0074bdb762c39dbb986c":{"kind":"REWORK","_number":1,"created":"2020-08-17 22:29:48.000000000","uploader":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"ref":"refs/changes/12/746612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/746612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/746612/1"}}},"commit":{"parents":[{"commit":"7cbc161067d2b18a0c428a9bc2e6c42ec1022e23","subject":"ironic-pxe: add syslinux-tftpboot on ppc64le","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cbc161067d2b18a0c428a9bc2e6c42ec1022e23"}]}],"author":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:19:49.000000000","tz":-240},"committer":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:29:37.000000000","tz":-240},"subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","message":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\n\ngrub2-efi-aa64-modules ships in RHEL and CentOS on all GRUB\narchitectures (x86_64, ppc64le, aarch64, but not s390x), and therefore\ncan be used on any of them to support heterogeneous clusters with AArch64.\n\nChange-Id: Icf7e7bfddec89c4aec89410884480539b6bbfd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e2421a1de1259d751d0074bdb762c39dbb986c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e2421a1de1259d751d0074bdb762c39dbb986c"}]},"branch":"refs/heads/master"},"b3c8e42ac2bc0392efe992f4329bb481633b72f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-24 13:29:43.000000000","uploader":{"_account_id":32322,"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","username":"yselkowitz"},"ref":"refs/changes/12/746612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/746612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/746612/2"}}},"commit":{"parents":[{"commit":"0e756bfd6546590656bece95fc0cb3026798b840","subject":"Merge \"Add glusterfs-fuse package to manila-share container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e756bfd6546590656bece95fc0cb3026798b840"}]}],"author":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:19:49.000000000","tz":-240},"committer":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-24 13:29:43.000000000","tz":0},"subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","message":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\n\ngrub2-efi-aa64-modules ships in RHEL and CentOS on all GRUB\narchitectures (x86_64, ppc64le, aarch64, but not s390x), and therefore\ncan be used on any of them to support heterogeneous clusters with AArch64.\n\nChange-Id: Icf7e7bfddec89c4aec89410884480539b6bbfd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c8e42ac2bc0392efe992f4329bb481633b72f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c8e42ac2bc0392efe992f4329bb481633b72f0"}]},"branch":"refs/heads/master"},"f5d1caed9d714a65bc865f32efaf5a7f52a26676":{"kind":"REWORK","_number":3,"created":"2020-08-31 08:57:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/746612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/746612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/746612/3"}}},"commit":{"parents":[{"commit":"5dcd1dd915189e047e1e3b481c1b0d2dba9ea2fe","subject":"Merge \"Fix kolla_set_configs --check with a directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dcd1dd915189e047e1e3b481c1b0d2dba9ea2fe"}]}],"author":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:19:49.000000000","tz":-240},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-08-31 08:55:54.000000000","tz":120},"subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","message":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\n\ngrub2-efi-aa64-modules ships in RHEL and CentOS on all GRUB\narchitectures (x86_64, ppc64le, aarch64, but not s390x), and therefore\ncan be used on any of them to support heterogeneous clusters with AArch64.\n\nChange-Id: Icf7e7bfddec89c4aec89410884480539b6bbfd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5d1caed9d714a65bc865f32efaf5a7f52a26676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5d1caed9d714a65bc865f32efaf5a7f52a26676"}]},"branch":"refs/heads/master"},"3e0cf599f784adbfc5fde774492816c809f66d78":{"kind":"REWORK","_number":4,"created":"2020-08-31 09:38:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/746612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/746612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/746612/4"}}},"commit":{"parents":[{"commit":"5dcd1dd915189e047e1e3b481c1b0d2dba9ea2fe","subject":"Merge \"Fix kolla_set_configs --check with a directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dcd1dd915189e047e1e3b481c1b0d2dba9ea2fe"}]}],"author":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:19:49.000000000","tz":-240},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-08-31 09:38:14.000000000","tz":0},"subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","message":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\n\ngrub2-efi-aa64-modules ships in RHEL and CentOS on all GRUB\narchitectures (x86_64, ppc64le, aarch64, but not s390x), and therefore\ncan be used on any of them to support heterogeneous clusters with AArch64.\n\nChange-Id: Icf7e7bfddec89c4aec89410884480539b6bbfd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e0cf599f784adbfc5fde774492816c809f66d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e0cf599f784adbfc5fde774492816c809f66d78"}]},"branch":"refs/heads/master"},"c3a3c7bee0d5fd68ceef1446a43f19b2d4512614":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-06 09:05:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/746612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/746612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/746612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/746612/5"}}},"commit":{"parents":[{"commit":"2725ed9d842622078e31165d91718e58d07620e0","subject":"Emergency fixes to pass CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2725ed9d842622078e31165d91718e58d07620e0"}]}],"author":{"name":"Yaakov Selkowitz","email":"yselkowi@redhat.com","date":"2020-08-17 22:19:49.000000000","tz":-240},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-06 09:05:27.000000000","tz":0},"subject":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures","message":"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\n\ngrub2-efi-aa64-modules ships in RHEL and CentOS on all GRUB\narchitectures (x86_64, ppc64le, aarch64, but not s390x), and therefore\ncan be used on any of them to support heterogeneous clusters with AArch64.\n\nChange-Id: Icf7e7bfddec89c4aec89410884480539b6bbfd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3a3c7bee0d5fd68ceef1446a43f19b2d4512614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3a3c7bee0d5fd68ceef1446a43f19b2d4512614"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
