)]}'
{"id":"openstack%2Fkolla~747507","triplet_id":"openstack%2Fkolla~stable%2Frocky~I80249e53f9cd09a2abd1178a8f50f750b957b210","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I80249e53f9cd09a2abd1178a8f50f750b957b210","subject":"[DO NOT MERGE] Test CI","status":"ABANDONED","created":"2020-08-22 02:26:08.000000000","updated":"2020-09-04 07:37:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"126d76339c537deb7233329413cc14678f7b1cff","_number":747507,"virtual_id_number":747507,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-22 08:12:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-22 04:22:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-22 04:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-22 08:12:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"0bf6e2d0c83badd247a66700711d384a128e7065","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-22 02:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcbf1e6863022409ea7c196c2fa7f7ebe716e168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 04:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3e074bd23f854f9f81484bbae88a403d : SUCCESS in 36m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bdd0f9912513451f99cd15cefa0979ae : FAILURE in 50m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/32c5e9d8544d4dfe99bdeba301065766 : SUCCESS in 1h 05m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1b7ecad0470a4717a31e9e59efacd2df : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/445b77692dbb4556940c3e76e4eca914 : SUCCESS in 50m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dc9b72c8760549709c8feb9b60ca4f0a : FAILURE in 52m 10s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e90262c205c4457d9c5bc5e243768645 : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4d7aa569f3c44ce6bd16a772a548a546 : SUCCESS in 1h 12m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9396f29779cb455c9b19a99edbfed1c3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c033a8a32744d0ba1352b522786d527 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47ec5f4d97014011bb9069915f8fb08e : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92373cfaafc644ac89789b18acb56799 : SUCCESS in 5m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ddc77d247d8045f7897f28ef61d35620 : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/724ee133fff440e9ac6ee60ec2e392db : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ede4175fc9a34746947584dd1e409b07 : FAILURE in 34m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74087c88194e44b98326d1ce00bbb5c4 : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/36302c8cabf14a8cabe4ca9d372854cd : SUCCESS in 1h 51m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"4eb5382d7a6e97748983526f5bdc2b62098bf3e4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-22 08:12:03.000000000","message":"Patch Set 1:\n\nyeah, it\u0027s b0rken","accounts_in_message":[],"_revision_number":1},{"id":"126d76339c537deb7233329413cc14678f7b1cff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:37:53.000000000","message":"Abandoned\n\nshould be fixed by https://review.opendev.org/749881","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bae6d08164cc577b57b08ccfdeb1bd6be63efaf1","revisions":{"bae6d08164cc577b57b08ccfdeb1bd6be63efaf1":{"kind":"REWORK","_number":1,"created":"2020-08-22 02:26:08.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/07/747507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/747507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/747507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/747507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/747507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/747507/1"}}},"commit":{"parents":[{"commit":"5b64da9f319162c9d17e2894d8f5b0658d661e75","subject":"Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b64da9f319162c9d17e2894d8f5b0658d661e75"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-22 02:25:43.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-22 02:26:08.000000000","tz":0},"subject":"[DO NOT MERGE] Test CI","message":"[DO NOT MERGE] Test CI\n\nChange-Id: I80249e53f9cd09a2abd1178a8f50f750b957b210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bae6d08164cc577b57b08ccfdeb1bd6be63efaf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bae6d08164cc577b57b08ccfdeb1bd6be63efaf1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
