)]}'
{"id":"openstack%2Fkolla~747510","triplet_id":"openstack%2Fkolla~master~I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388","project":"openstack/kolla","branch":"master","topic":"add_glusterfs_driver","hashtags":[],"change_id":"I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388","subject":"Add glusterfs-fuse package to manila-share container","status":"MERGED","created":"2020-08-22 03:00:47.000000000","updated":"2020-08-24 12:14:03.000000000","submitted":"2020-08-24 12:04:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747510-1598270663010-038e877c","meta_rev_id":"c533e378817245943ce049fcc599b913a5c95048","_number":747510,"virtual_id_number":747510,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 12:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-08-24 10:39:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-08-24 09:23:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-08-24 10:39:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 09:23:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-24 10:39:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-24 12:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2d061e0caa3871f33a48d73553f4e0eba08b248","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-22 03:00:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd27d687ad8c7113c43dc58c4c31849f83cf575a","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-22 03:08:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e87321bd1abd9cfb5d1427dd6176570264c244df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 04:29:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/52826106284e4aba8f4492cdeb0ac501 : SUCCESS in 38m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/081adb2a96fd471e940ca6bed7cc2f7e : SUCCESS in 28m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/299a13c9324e48c681c8a4d9c73489d2 : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c167743a1a0844ddb99982836a408180 : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3542d1b9fdb458ebcb76c26a39b1438034aaf41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 04:34:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b678b081a09c42d6a4293cdd2f458eee : SUCCESS in 21m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c539426ae5854d8ab755d34633d44b24 : SUCCESS in 23m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e98b353413af46e8926fab7f1688e5db : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/976993f96da34efd9111bfddf49e4ce9 : SUCCESS in 1h 23m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/830b953e550e4ab4b08957dcbfb97fa7 : SUCCESS in 1h 13m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fc7c10a960f842a7a08db07d239cec60 : SUCCESS in 46m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3614c6af9e94df6bae95d39020dfa20 : SUCCESS in 16m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0ad93d54fbd47b0bb04d5534c6a6ae1 : SUCCESS in 15m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c604e1fa4f5b49fc85f22672d34e485a : SUCCESS in 57m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15ac967509424b1d86516649cc7eef13 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b872aae3a0f4b6d884508f63990d265 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d7316bb620d4609b55bb7882bf8a2f9 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/debb38e384eb4f6f82b68d4b31c51b31 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918c89ff3e984d6589e8f32944e7b122 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e8d1ac065f84acc95b5278f7567ca2f : SUCCESS in 19m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f473f06d573b4a7f8bdde02a525e3cb6 : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82fb8a90b02545e2b86d295a983a40f4 : SUCCESS in 53m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a8418ac04454e23bdc0fe341f23e012 : SUCCESS in 52m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/859a33d9062e4d5896c1ecee527d595d : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ffacb2afd1a7cde6b496e7c0a66045ad7007cc63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-24 07:25:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5b8a2fa382359de5bda78c2464c274d46e0d279d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 09:18:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fda8f8e731dd35f5905d33a6de743e3e9d4e852f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-24 09:23:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"36e44408d416c8e2d118ec1fab77ce9906df48cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 10:39:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/10fc373486d34f16a48094fa2283f528 : SUCCESS in 25m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a59da8112da64a83840422a71045a625 : SUCCESS in 26m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/67b6cddf36884542a7c7c7eee3282fb2 : SUCCESS in 59m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/409d5c3657314a08a1bb82c840f4ebdb : SUCCESS in 58m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a24981f2586e4d5b8c76c45fbca94f14 : SUCCESS in 1h 14m 58s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/680a70b1e4904cf69d8c9a01e02f7759 : FAILURE in 53m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/803720ecef1841c58375f2fec2a308fc : SUCCESS in 16m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b5d1f332a67c40008e6618190e6ef0e4 : SUCCESS in 16m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ff15c70e8d54d82949963dbeb2afbd0 : SUCCESS in 1h 00m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdef8ca7898b4546b0e6374f19587c88 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038102eca0da455bb902acc14e871251 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eba31b19a134d32ba38be8b6b548e7e : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1a0962898714515922bcc80e3f72407 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f95ac6d42e9474c943c2bcef2c36290 : SUCCESS in 7m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea413d66d8e949ccbcf447097856c4d4 : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b708df69f7234a829a82caead326ec68 : SUCCESS in 15m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19f53e4de7c94c2294a59da31d7c8bb4 : SUCCESS in 48m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cd588c931e54b769871081cf2d6b653 : SUCCESS in 42m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/925eb8c421df4ee3905364ea61e7f8d9 : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":3},{"id":"4f856d9a2ad6ff8fa48df2b07d1419e8057ca596","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 10:39:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"986e7d3ad155aa1333dcea64fd6db67a2842df2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 10:40:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1499b1192404e10b9688271c7b40ade6472661b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:18:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e5bf31798064b48a36368af42767ce8 : SUCCESS in 38m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1576570baf244629360cc2fe4641edb : SUCCESS in 31m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d20beb75b426448fa1669505cda27b03 : SUCCESS in 1h 53m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/46ed3fd7527740c5a90dea36fd645289 : SUCCESS in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"59dee209371f5514aeb093b4c740ab7883559989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:04:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/16083c78c52843dd9a58ff76b160efea : SUCCESS in 23m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d261729aae5244648655de09693fbbb7 : SUCCESS in 1h 18m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e7a42be95ebc472f9230ec415e2579b7 : SUCCESS in 1h 21m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9569bed295645589837675203bf7137 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6f02878bd14d799f9ecd9baa2801d1 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47552b5712694e6894834f9c63d410a3 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309cd30d039948e29ab4310b1f238e5f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fcf0919d1a64a42955fa63895e64d68 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9376f6957fdc4d4a8b3264a380d96b9f : SUCCESS in 22m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d8baec62107439d977b6467fc343690 : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc5f96b13f474f45b158509348401770 : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":3},{"id":"7cbb4672d7826ebd8bf7cbd67aed25e185bda498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:04:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"633bbce3694af45e0b223694e4772bbf231c03a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:06:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533e0ca9675e49b3899b927624144e6d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3},{"id":"657a47bb19692e41e20be9b5a04c0b0ee3968397","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:12:37.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 821c49600a4cdde9d3b2afdf1deb8bbcca310637","accounts_in_message":[],"_revision_number":3},{"id":"eab3c06a993820f72bb0fed066c45371f0cf354a","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:13:00.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 29b9b87fa5feab53329334c570694cbb521b7815","accounts_in_message":[],"_revision_number":3},{"id":"93e7f8300f554fd5ccef9d7a6845f14012dff568","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:13:18.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 05de98df229ea047fd794ece869bae2d0c19817c","accounts_in_message":[],"_revision_number":3},{"id":"635404f062834b71c717cdb59c3e52a9354357d7","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:13:36.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 4af78543f077738e933c6344cf218fd5aad89d66","accounts_in_message":[],"_revision_number":3},{"id":"10fcc776d9bee42334dde997b69ec0ba023f71fd","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:14:03.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit ee23dcf5adb881469b1766aae7dd715a24fce5c0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9ca43b9ed7814045a0219e17524f4e2aed2f1771","revisions":{"10904ce9b04ef09aa28f9603cf2a45647f8e5201":{"kind":"REWORK","_number":1,"created":"2020-08-22 03:00:47.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/10/747510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/747510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/747510/1"}}},"commit":{"parents":[{"commit":"e43ec971bc7a4ae42ddeee43248254cf7c26d3f4","subject":"[doc]remove sensu from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e43ec971bc7a4ae42ddeee43248254cf7c26d3f4"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 02:57:00.000000000","tz":0},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 02:57:00.000000000","tz":0},"subject":"Add glusterfs-fuse package to manila-share container","message":"Add glusterfs-fuse package to manila-share container\n\nChange-Id: I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10904ce9b04ef09aa28f9603cf2a45647f8e5201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10904ce9b04ef09aa28f9603cf2a45647f8e5201"}]},"branch":"refs/heads/master"},"a1fd5729ef132d3d950af5cd756d2364158f2f9f":{"kind":"REWORK","_number":2,"created":"2020-08-22 03:08:25.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/10/747510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/747510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/747510/2"}}},"commit":{"parents":[{"commit":"e43ec971bc7a4ae42ddeee43248254cf7c26d3f4","subject":"[doc]remove sensu from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e43ec971bc7a4ae42ddeee43248254cf7c26d3f4"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 02:57:00.000000000","tz":0},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 03:07:54.000000000","tz":0},"subject":"Add glusterfs-fuse package to manila-share container","message":"Add glusterfs-fuse package to manila-share container\n\nChange-Id: I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1fd5729ef132d3d950af5cd756d2364158f2f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1fd5729ef132d3d950af5cd756d2364158f2f9f"}]},"branch":"refs/heads/master"},"9ca43b9ed7814045a0219e17524f4e2aed2f1771":{"kind":"REWORK","_number":3,"created":"2020-08-24 09:18:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/747510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/747510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/747510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/747510/3"}}},"commit":{"parents":[{"commit":"e43ec971bc7a4ae42ddeee43248254cf7c26d3f4","subject":"[doc]remove sensu from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e43ec971bc7a4ae42ddeee43248254cf7c26d3f4"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 02:57:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-24 09:18:36.000000000","tz":0},"subject":"Add glusterfs-fuse package to manila-share container","message":"Add glusterfs-fuse package to manila-share container\n\nChange-Id: I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ca43b9ed7814045a0219e17524f4e2aed2f1771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ca43b9ed7814045a0219e17524f4e2aed2f1771"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
