)]}'
{"id":"openstack%2Fkolla~747695","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388","project":"openstack/kolla","branch":"stable/ussuri","topic":"add_glusterfs_driver-stable/ussuri","hashtags":[],"change_id":"I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388","subject":"Add glusterfs-fuse package to manila-share container","status":"MERGED","created":"2020-08-24 12:12:37.000000000","updated":"2020-08-25 10:01:51.000000000","submitted":"2020-08-25 09:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747695-1598349538825-5554bd60","meta_rev_id":"18cd5be4a1cd855fb42711f084cf99cd24d6be59","_number":747695,"virtual_id_number":747695,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 09:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-08-24 19:34:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-08-25 08:23:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-08-25 08:23:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 19:34:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-25 08:23:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-25 09:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f95928e342e94ad335c81cce885140d63597103","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-08-24 12:12:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aa6abfe79dc7be2c304cdbdc4c9db0c8e73b3b67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 13:42:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bb569be7a0743229e803a56a5aeedeb : SUCCESS in 36m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/33aca58f04744a5dbf20ca577365d70f : SUCCESS in 33m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/166674bd8bc243e0ab2a9e5d8fc4a04b : SUCCESS in 1h 20m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf118fea60244983a02b958c6e84fd48 : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efc611f79aff35cd3a4bdf93357045dd6310b922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:27:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d2630ad6b8b34c6d800123e89b1f3d93 : SUCCESS in 2h 00m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fde6efa48b8c43f48b031e4b7ec059d3 : SUCCESS in 2h 00m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a5f02516aa748a98202868af8c65361 : SUCCESS in 1h 20m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a9703d5504f44d748721a34e815ae04b : TIMED_OUT in 2h 03m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/767fc748054d4ed58d32a1cc8ea7fb10 : SUCCESS in 1h 28m 40s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fa425a2eb29c46998e5e2223a489d21a : SUCCESS in 46m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83f93fb248374773b939bf1a7825206a : SUCCESS in 18m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b1e4e02fb904eef80d57fad2ce17403 : SUCCESS in 18m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/595e63c12c8548e3bf1fcd0463c19788 : SUCCESS in 57m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e9c97fe43844fd09a2659e927758cb3 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67d7d76b4384f5d9a0213ea8d6d62c7 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f56eebf5aca7464d98182b65e1ef14d5 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/418bd8f694924ba680c26b88646b2a3e : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae07d09adce483b9c004b2ad12aa52b : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2cd6d464dcee405d9c3802367c9eba36 : SUCCESS in 22m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/45165a84a1184c10ae56013a04f2a239 : SUCCESS in 25m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/38570fc9356f41a4be587c518a4d42a3 : SUCCESS in 45m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e76b05abc05240f2a6c2cf4ba9a356d8 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/429bbf100c0847aea79abc80e65684fd : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c16f99c0a2c658348c4cc835846ee6c4415ff330","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 19:34:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8690353cf0e19435165ca9698cac17ffa66d1e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-25 08:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30e2baf685322efc9adae083a4c037d6ba5fedb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:23:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da932c7c4ea2f97b6ff04da9e1b579d66f341c47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/177dd3bf07634564ae61e86a3ff25af0 : SUCCESS in 38m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ff9dceb740a24efb85c4e04d5dc65f47 : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/22990d320d734daebea32f2142bea9a7 : SUCCESS in 1h 08m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/281718d34a1248bfbe530369151d8b46 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984784f979d547aa940b89bc37a377eb : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951825082cef48f0928344cad6011a77 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fe28c5038fd4c009532c6420c95780d : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f955fb296bdd4c0db732e19a935453dd : SUCCESS in 9m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d53f3df0ca3404ba8d54da9e6b401c4 : SUCCESS in 34m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c94cd4f75d91406d9131ae7a170ae6bc : SUCCESS in 1h 16m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da9440199a1247248ea6733f89be1a87 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bb68da7ca875cacf0ef155d27467fd39ee563f9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:58:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e11d5d184e7b8c55063ab2de4aff85fc40f6da93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 10:01:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da2f59f114a46229593faaa5e4cc5e3 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"821c49600a4cdde9d3b2afdf1deb8bbcca310637","revisions":{"821c49600a4cdde9d3b2afdf1deb8bbcca310637":{"kind":"REWORK","_number":1,"created":"2020-08-24 12:12:37.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/95/747695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/747695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/747695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/747695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/747695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/747695/1"}}},"commit":{"parents":[{"commit":"69c0ff8491ac7b7f30410a2b3c1f575cec3c00b0","subject":"Updated TGTD to N for centos8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69c0ff8491ac7b7f30410a2b3c1f575cec3c00b0"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-08-22 02:57:00.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2020-08-24 12:12:37.000000000","tz":0},"subject":"Add glusterfs-fuse package to manila-share container","message":"Add glusterfs-fuse package to manila-share container\n\nChange-Id: I2b4cfdb25f3ff82d8d6fdebba8b75a818c10a388\n(cherry picked from commit 9ca43b9ed7814045a0219e17524f4e2aed2f1771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/821c49600a4cdde9d3b2afdf1deb8bbcca310637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/821c49600a4cdde9d3b2afdf1deb8bbcca310637"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
