)]}'
{"id":"openstack%2Fkolla~749454","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ia4b4c4c2a487e135b100038281ad212789592d14","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ia4b4c4c2a487e135b100038281ad212789592d14","subject":"Bump up versions for train","status":"MERGED","created":"2020-09-02 08:05:20.000000000","updated":"2020-09-03 22:26:49.000000000","submitted":"2020-09-03 22:24:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749454-1599171890292-33708db1","meta_rev_id":"a074fdff03299cce385ee98c1942afa447d157d8","_number":749454,"virtual_id_number":749454,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 22:24:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-03 18:12:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-03 18:12:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 18:12:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-03 18:12:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 08:46:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-03 08:59:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-03 18:12:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-03 22:24:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a15dbef913c98995d9329b19a5b50df2eac33e1f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 08:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e1434f41e38946468aeed22ebe672daea7f16ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:52:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5ed4d7be5ce84dc18417238a43b9180b : SUCCESS in 45m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/084cb93d51d4414999e646cb42dcd8ca : SUCCESS in 36m 21s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9ffb5d6b50f24b52a9580a31545d9d76 : SUCCESS in 22m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c6aad0a53db541ba8ab1c1c6954d39e0 : SUCCESS in 39m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ed40f325b4e14e4e9afb8b5d6b70ece8 : SUCCESS in 1h 24m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c7533b30329d4a4ba52acba0958d32fc : SUCCESS in 1h 10m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d5b9db2d4acd48cba693e9a77ce14475 : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a33dadbdd6e3419b895f24e394588823 : SUCCESS in 1h 45m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d35959bea9d94a8384055d4098b83e97 : SUCCESS in 1h 26m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c781735b84d45c78ce8f13c7ccdff44 : SUCCESS in 58m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0377e3312870432d9cd342a3330394ec : FAILURE in 12m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1fe1a6e196e543298e528126f08c89c3 : SUCCESS in 21m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5aa4a59a613b40e7a2029b612ec444eb : FAILURE in 13m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27170122bcaa47ba8052300a40e6e50d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3fe24e74be4a8d8d13b652f1866e27 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ebb8116114b4cf5815f057c91f2948a : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f894c1235dcc4df29db18ea813781e74 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7837af5100f94ce2a910a2dd44f1c59b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827020566d764b3cbdff8d06935235fb : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce8d2ca048fb4af690e1d4358402a7a0 : SUCCESS in 27m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33aa061eb1fc447799b3cf92fa8ca64f : FAILURE in 20m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba14ff1a89894662a97b7e20144bd091 : FAILURE in 14m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a32c78f37af647009d6964421430a594 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b24ec9c9f82347f7a71511ee73dea557 : FAILURE in 18m 45s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"c1fbe3bb52684a291b8068296738b6768c269dbc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 16:54:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"eeb1bd60bd48c1d9a64442c81df740cc4c8e6125","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:12:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f0ba1a4cd6d833c4b96653a04cc602cb9a89fdaf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:15:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1bdb7848b250b44e82fce834f30d88ff02491035","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:15:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c1cb831bd2bca4f7553e2a5b4e6ca2a3bc3b1ec0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 19:18:04.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ca4e75226033fa5de3433f06172223df81987122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:12:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8aad142bf19b463a94ed629734a93657 : SUCCESS in 54m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e64887cded23496b880472ae32a2d0ec : SUCCESS in 1h 03m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f79e22188d7045d9a30405f6f44a10e9 : SUCCESS in 49m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/216bf4b4e0864e52b8200c00efbd07a7 : SUCCESS in 50m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ac6e0121df2483baf5d5dfbd7250080 : SUCCESS in 1h 17m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8dbad9e0d1ad4043ab6111cb72ff5012 : SUCCESS in 1h 02m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6e53a6d493f44bd593ad6bfa35774143 : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9b6366a750c94c7ebae368111a42a4cd : SUCCESS in 59m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3d3cd31026034690b618b55bf969d536 : SUCCESS in 1h 32m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8886d12db34a46dc9be658149b6f96b1 : SUCCESS in 1h 02m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dcc7ce988b614203880b8cdeb4777437 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7324b70f9ddd4cf8ab7b1957c8150f36 : SUCCESS in 36m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8bed58c5e7ab4e778f3c7027c8593622 : SUCCESS in 50m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25b2461dbfdf4e2ca1d730e5632bdf43 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be08d37e0cd48bf8f855b90c3004c61 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f31ab27633f347f9b4ff509b14f8df74 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03a034b0384a40f78e2e3ea983ffa1f0 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18d666f9ec234efaa1114c6fa984df9e : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82b41ac22d24e338c8fa4265381304a : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7887279657244eecbc04399eefe0bcf6 : SUCCESS in 24m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/50edb48c7a4a42cf93851c5b2cb5b7a2 : SUCCESS in 23m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdaea391cbbe4f18aa6ad3429b348fdb : SUCCESS in 48m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff4e0fb1f2c8416cb81ea1aa9fe1acf6 : SUCCESS in 49m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/25e331cd95484e76a47894fc3f2d8458 : SUCCESS in 1h 36m 47s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"6fcfc4949e509c545be032057bef2e376b94df48","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 07:01:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5fa372bc7ffbcde654dd64fd926ae5711fb45691","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-03 08:46:36.000000000","message":"Patch Set 6: Code-Review+1\n\nThe updated Nova release includes the fix for OSSA-2020-006.","accounts_in_message":[],"_revision_number":6},{"id":"c9032bff7a2eae099f9e1945a44e8d1899e3b3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 08:56:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/14d0e8975c4b417ca9a38342c205cc6a : SUCCESS in 43m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2550d484c60641a9ac261cc2fb393939 : SUCCESS in 1h 13m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9294725af3224d4cac8a632d2d312601 : SUCCESS in 24m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8227d6a050a14c05ac1b9418d66e1913 : SUCCESS in 31m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/372c16247b3240468a35cfd64f8a4646 : SUCCESS in 1h 22m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41349c06de634dd48097d18cbf756be5 : SUCCESS in 1h 10m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/25fff9689cd54641b71941eee5e9f735 : SUCCESS in 1h 17m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8859604efae64bfb889e65390f167054 : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5d6302a872724b63a60d2d15afef2863 : SUCCESS in 1h 18m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d98c0735470849a1b32f4074f0981ad4 : SUCCESS in 1h 00m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fbb6296cf0e44deaa5991821e0b805a4 : SUCCESS in 20m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/78ca7aa1dc124871ad60afb15cbfb77d : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f545278e822497b8c947d8383ed142a : SUCCESS in 52m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d24472b61534ce7a731d59f0979be93 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03bcd0ab0f041648e21ab452c5cd8dc : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f4b72ea84de439b914b56279e083077 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b09a8814283745ebab023f230650b2be : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c30fdccfdb5e4f3394403f6e6407d455 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c124c000fe534c70adf1e0813c6308c6 : SUCCESS in 8m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b0b4eef74e1475ea111ef74f68af9dd : SUCCESS in 22m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8715b779d56a4547b3f953bb0aa55302 : SUCCESS in 44m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/766c8336e1b74452b950b6772cdee542 : SUCCESS in 1h 05m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b72363205d35495b97284d692b4eeed3 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21ebf08a259d42a38206267c46816720 : SUCCESS in 1h 15m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"96efc0abbe91cd892d787df6ab773205dc160179","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-03 08:59:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"dfade78bc53eda8cd4c23fad461b12669493437b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:57:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"79663ba53b9bae525ee8959ce00f80309b11cc9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 13:00:02.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"d3086c26e8a0f5e62549c7547330d1ba5968cbf0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"373c2b159b7936ca37b05ece9d1cf9da9291d6ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:12:54.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"bce31cea34fc8551bd78442d2125b41283f3a636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:06:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed9df5a1de45427892d562a8a7d3b7fe : SUCCESS in 43m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e90f7d1ae8454214bef7b41946492c4c : SUCCESS in 43m 18s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3fcbb0e17f4f4e4092447c03c5fc729d : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7e426fba24a44a9e8a5fe31cc5caf07b : SUCCESS in 1h 42m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67118ca81a4b428793563ed90385282f : SUCCESS in 1h 09m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/feccfc70dba545b0bd5d0b585e0e3eb1 : SUCCESS in 55m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fa618a059952481b803f7ba1354d45e9 : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e163cf302fd142b7a3d6ee98789a6b77 : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d5b88cc00f9a4d42be84bf910d3098d3 : SUCCESS in 1h 49m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/183ed0ea177549bdbb30f6ba1f11a8df : SUCCESS in 55m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7d0c4f18adc24a58a60b43551079a1a5 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63db2f721b5a42d88de6af86e4b66dd0 : SUCCESS in 35m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4ad69c3a7e274cffb4f2760747b5eb25 : SUCCESS in 48m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89fdcdb9bc774a00a475e5daa7250987 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874dd0dcb4dc42cdba00449b893a6fa9 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00c013b861b84e21b8ef3eaa3a224b6f : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e948793039b54959a606da45c80f6e99 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c96f6a30151d4f84af68433f592f3e7b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57e21315a0d40ddb674ca99c437a2a1 : SUCCESS in 8m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1696ffb97ea748feb2c040ae15a0ccf3 : SUCCESS in 1h 28m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f1b8e05a4564b53a3855f7445ffc505 : SUCCESS in 24m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fabfa5a04c3e459f8040f5e170cf44dd : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/321c03756a264deab9a228b45f4cff93 : SUCCESS in 47m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/186160db213545eea3a7f16fdd381283 : SUCCESS in 1h 14m 59s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"b6c17f5e761f1a5af25d7f596e206052cdc56b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:54:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6b0a39bfccf2e3957d3b74260e434ccd99b38c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:24:50.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eeb8140faf6a46149eca703807fba526 : SUCCESS in 1h 02m 08s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/eda3a7c4aa5a4278a8446de5c5ba2486 : SUCCESS in 42m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e8df6ef380174ab9bdf0427937234720 : SUCCESS in 1h 17m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bcd540503fdb4620a807b4f851ea176e : SUCCESS in 1h 03m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/49ba96d872a042279194e228bc0572e6 : SUCCESS in 1h 22m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d3e31495dd874f7395ae8f2d45ab6e59 : SUCCESS in 1h 00m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d15ba9001f074866a58f4082f54c1ac8 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd63e932757544909c40770ea75422d2 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c63471d46884493b2a3bc8200a0dd2b : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b93ad9d8d54148748453c3018aab3635 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f56ea49ee31d4dc1a893e74ebdad04c8 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08be21ff2a814c0fb2d56d3db9d8bf6d : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2d6ca5c4ace4e4c94606628a45a1103 : SUCCESS in 23m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85bff1dc9d3b4a80815120e18db20ec9 : SUCCESS in 49m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d8d24abbbda4955ae7f8f6886f0dade : SUCCESS in 1h 00m 42s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"bc412c2b6affb604e89fbffaea2c57d8cb4b620d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:24:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"a074fdff03299cce385ee98c1942afa447d157d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:26:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d870032d27014dcfbe2c913b37b23fa3 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2ff242c4a270258fed0a0dbdbd3b66a402cadece","revisions":{"42a0c6e254add1e12359405f64f48be3f243a8a9":{"kind":"REWORK","_number":1,"created":"2020-09-02 08:05:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/749454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/1"}}},"commit":{"parents":[{"commit":"65d9a041723d60ccea4f1ffa39134beb4938ea9d","subject":"Updated TGTD to N for centos8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d9a041723d60ccea4f1ffa39134beb4938ea9d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:05:17.000000000","tz":120},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a0c6e254add1e12359405f64f48be3f243a8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a0c6e254add1e12359405f64f48be3f243a8a9"}]},"branch":"refs/heads/stable/train"},"f8208f6b5b330f59dd1aedcdc2cb9393a05a79f1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-02 16:54:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/2"}}},"commit":{"parents":[{"commit":"1a7737fd18d4b3f183151a27535c176840e88282","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a7737fd18d4b3f183151a27535c176840e88282"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 16:54:10.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8208f6b5b330f59dd1aedcdc2cb9393a05a79f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8208f6b5b330f59dd1aedcdc2cb9393a05a79f1"}]},"branch":"refs/heads/stable/train"},"e08c1882f5543d05e61a72d7b2f1c3aad988852f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-02 19:12:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/3"}}},"commit":{"parents":[{"commit":"4638f761a6413189dea9dcd0b2c3950422a58cd4","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4638f761a6413189dea9dcd0b2c3950422a58cd4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:12:10.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e08c1882f5543d05e61a72d7b2f1c3aad988852f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e08c1882f5543d05e61a72d7b2f1c3aad988852f"}]},"branch":"refs/heads/stable/train"},"6dca929e2aca7f1bfb7ed003c869a4aac555c7e3":{"kind":"REWORK","_number":4,"created":"2020-09-02 19:15:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/4"}}},"commit":{"parents":[{"commit":"4638f761a6413189dea9dcd0b2c3950422a58cd4","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4638f761a6413189dea9dcd0b2c3950422a58cd4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:15:10.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dca929e2aca7f1bfb7ed003c869a4aac555c7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dca929e2aca7f1bfb7ed003c869a4aac555c7e3"}]},"branch":"refs/heads/stable/train"},"e57a177243a532cfcdc3179929cdb01241aa78b9":{"kind":"NO_CHANGE","_number":5,"created":"2020-09-02 19:18:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/5"}}},"commit":{"parents":[{"commit":"6cefed0ab64ca9776ee3c8274162ccf79cdd8fed","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6cefed0ab64ca9776ee3c8274162ccf79cdd8fed"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-02 19:18:04.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e57a177243a532cfcdc3179929cdb01241aa78b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e57a177243a532cfcdc3179929cdb01241aa78b9"}]},"branch":"refs/heads/stable/train"},"07f101611b5415753c05c7bb45fb6a8438956094":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-03 07:01:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/6"}}},"commit":{"parents":[{"commit":"326823c4d3661ee5cc329857eab65148aa1d8fba","subject":"Revert \"openstack-base: pin setuptools \u003c 46 to get horizon working\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/326823c4d3661ee5cc329857eab65148aa1d8fba"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 07:01:59.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07f101611b5415753c05c7bb45fb6a8438956094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07f101611b5415753c05c7bb45fb6a8438956094"}]},"branch":"refs/heads/stable/train"},"2ff242c4a270258fed0a0dbdbd3b66a402cadece":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-03 18:12:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/749454/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/749454/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/749454/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/749454/7"}}},"commit":{"parents":[{"commit":"a31836265efc6d42640dd7bc97bce0113441fb5c","subject":"[CI] Fix wheel url for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a31836265efc6d42640dd7bc97bce0113441fb5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 08:04:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 18:12:32.000000000","tz":0},"subject":"Bump up versions for train","message":"Bump up versions for train\n\nChange-Id: Ia4b4c4c2a487e135b100038281ad212789592d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ff242c4a270258fed0a0dbdbd3b66a402cadece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ff242c4a270258fed0a0dbdbd3b66a402cadece"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
