)]}'
{"id":"openstack%2Fkolla~749469","triplet_id":"openstack%2Fkolla~master~Idf401536b482b4088f5fb66408917752fd71726c","project":"openstack/kolla","branch":"master","topic":"fix_bifrost_network","hashtags":[],"change_id":"Idf401536b482b4088f5fb66408917752fd71726c","subject":"Fix bifrost network config","status":"MERGED","created":"2020-09-02 09:14:35.000000000","updated":"2020-09-02 18:15:21.000000000","submitted":"2020-09-02 18:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749469-1599070388920-1545f216","meta_rev_id":"fb5eaef8e1c936ca55b974f7ad7d68d857519724","_number":749469,"virtual_id_number":749469,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 18:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-09-02 15:25:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-02 15:24:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-09-02 16:51:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-09-02 18:15:21.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2020-09-02 15:24:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-02 15:24:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 13:44:56.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-09-02 15:25:20.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-02 16:51:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-02 18:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4d479aeb29ed1e4af0415f7193794cc56b3384e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 09:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bea0a4f2d9f22b2ee7debc59b02879b2ab8f7ad3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 09:24:39.000000000","message":"Patch Set 1:\n\nbut we are using venvs - are we hit by it now?","accounts_in_message":[],"_revision_number":1},{"id":"8ed9226d77a71f9f9a5ca1cfd7f8dbe49ea816e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 09:25:21.000000000","message":"Patch Set 1:\n\noh, it seems we are, odd","accounts_in_message":[],"_revision_number":1},{"id":"88da20ee93bffcb00a0d7414edf295f20fa5b41a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 09:27:00.000000000","message":"Patch Set 1:\n\nwell, then I don\u0027t know; is setuptools 50 just broken by design for some deps? It was supposed to break only editable installs and only against the system library and we do neither...","accounts_in_message":[],"_revision_number":1},{"id":"126bfeb8a0c54beeaf39734cb49e0889799c06eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 10:20:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18307b4fa2f76307239210f2a59ec5db0d192704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:44:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb24539da27248b7a585aabb9681a7ee : SUCCESS in 36m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5486523897e94d7bbc252555d41a3439 : SUCCESS in 30m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb319e5c74ec4844960592d010a69966 : SUCCESS in 1h 18m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ce367b24c414b4c8e824dcd091755ed : SUCCESS in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f8dd1f0e5c7ca5508c54c0f22502276fd22f926","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 12:04:02.000000000","message":"Patch Set 2:\n\n\u003e well, then I don\u0027t know; is setuptools 50 just broken by design for\n \u003e some deps? It was supposed to break only editable installs and only\n \u003e against the system library and we do neither...\n\nWell, it\u0027s broken on Ubuntu builds, funny to say that it\u0027s breaking openstack-base in Train and kolla-toolbox in Ussuri, and nothing in master.","accounts_in_message":[],"_revision_number":2},{"id":"2275cc6e7d6b918544d8ae62e695ce4d113c6123","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-02 13:03:18.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ccc3327ef734cf341a51b13724a81fb83d13242b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-02 13:44:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"db9b681629be041e29d0f7ea35f07bfed37a0b23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 14:07:34.000000000","message":"Patch Set 2: Code-Review-1\n\nsorry, it does not make sense; the setuptools is setuptools-45.3.0; also - the same code is in kolla-toolbox but it works there; at least the commit message is wrong with the accusations...","accounts_in_message":[],"_revision_number":2},{"id":"3afcc3f2dffbc0393b2ff8ea03b9e1466ed69fbb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 14:08:01.000000000","message":"Patch Set 2:\n\nI guess the culprit must be ubuntu itself","accounts_in_message":[],"_revision_number":2},{"id":"033590972748d4c40fbe090419c7fce5a67d8cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:10:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d24aabd370c248c4bfd33513d15e901f : SUCCESS in 45m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b869952c7aa94a93aa5e70d7d2371e39 : SUCCESS in 44m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/05152d8959624e0badbd1957ffc1d015 : SUCCESS in 1h 12m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ad7b1bb18fb9499facb4aeb7c056198b : SUCCESS in 1h 46m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6aed42a484554e37966c98eb7f18c9ed : SUCCESS in 1h 53m 37s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/71f3fd13d3db4f5bbebdb5ef76a80cf2 : FAILURE in 1h 09m 06s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/e35d2167fbb04f05a733123b06339ea6 : SUCCESS in 32m 51s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6584047aae91420ba28f7a2c9ddccde5 : SUCCESS in 1h 42m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5367a4b753fa48758fdbda685e674a0d : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/9b3c01cd515c471abd1a73d129466531 : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7208e2e6c2934465b48b81525453c569 : SUCCESS in 1h 35m 14s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/821621ac7d9f4ea5889556bc7577fd93 : SUCCESS in 1h 05m 50s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/0aff4ecf14b4458289af900a2bba7a4e : FAILURE in 1h 43m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9a21358cf7a422586cc9c249887a405 : SUCCESS in 17m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/00658098dd1a473089503657f0a0d439 : SUCCESS in 17m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a64be088abf340a5b3e28200fff311ae : SUCCESS in 55m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e34f60780a954f77b0f3edd473670fb4 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16cc8364e234e9fb11c86073133929b : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84b8d00e03934da8bd65f529f98b0226 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/642e9508a27043b89bfe140a7d4bb0c5 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ba318fc8f04243a436c3dec073592f : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5bc8160372b043ad9ee6c908ed893e13 : SUCCESS in 20m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd282cad43f94e96a82a60afab80aa5e : SUCCESS in 21m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c63b5b89c4a4e6595cf38b511055366 : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/059e66ffe1d84b70a0a81742e28c9b49 : FAILURE in 30m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/68e9249f4ede4abeb0a8a897c5c146d5 : SUCCESS in 1h 14m 09s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8b460b3cfdbc40c18e9721852317c51d : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/db7eeb96e9124e349e962fe7260c2099 : POST_FAILURE in 1h 00m 35s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f8cb4e2fb4b14dd8a7baf4e8642a0bc4 : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c65350029b1c4534980a8594f40ee010 : SUCCESS in 52m 12s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8682d4c8a4d245b39bda3320bb7ffd9a : SUCCESS in 1h 13m 03s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/36ca8ebd2df749e3afb338a91c75d799 : SUCCESS in 54m 34s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/f032c22087d7401b8abf756be8c615cf : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/0d1060b4af50451ab6be81d2682a6943 : SUCCESS in 1h 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b00a907cccb90a04b3c105111787224570352c34","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 14:27:39.000000000","message":"Patch Set 2:\n\nok, so in kolla-toolbox we are doing (expanded):\n\n  {{ macros.install_pip([\u0027virtualenv\u0027]) }}\n\nbut in openstack-base:\n\n  {{ macros.install_pip([\u0027virtualenv\u0027], constraints \u003d false)}}","accounts_in_message":[],"_revision_number":2},{"id":"f1f8a1937914f55b0d51f561c03ea5fe48f3b8cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 14:35:10.000000000","message":"Patch Set 2:\n\nmaster is not failing on setuptools...","accounts_in_message":[],"_revision_number":2},{"id":"f3d99ddd8cac129c310b12be538844c43f97b4ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 14:35:27.000000000","message":"Patch Set 2: Code-Review-2\n\nblocking this, please repropose bifrost fix","accounts_in_message":[],"_revision_number":2},{"id":"c4f23ee9d909c3422ccdb977b9a98475a6ff549f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 15:23:19.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"65f22f1d75d220449c355571792596791ef336cb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-02 15:23:31.000000000","message":"Topic changed from fix_setuptools to fix_bifrost_network","accounts_in_message":[],"_revision_number":3},{"id":"7d036f786e13618061f42dd6ce536299828bb2f9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 15:24:21.000000000","message":"Patch Set 3: Code-Review+2\n\neh, I thought you would use a new change id","accounts_in_message":[],"_revision_number":3},{"id":"c039d6f9add763d129f1922ff79b01fd3d4a8a3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 15:24:33.000000000","message":"Patch Set 3: Backport-Candidate-2","accounts_in_message":[],"_revision_number":3},{"id":"8027a3e65e652e356da10dda659ec443dabf7235","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 15:24:40.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"d1d1186e66874d640807bb8b35197acaa720870e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-02 15:25:04.000000000","message":"Removed Backport-Candidate+2 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"c267f9e6459ca1642b0476ad89ca8a1b50ef8fc7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-02 15:25:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7d424e1c0e9ab761a745d569313ae3fbf7bb2a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:51:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e4fa17e5795e4274b8e293673fc21d93 : SUCCESS in 36m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0c3c73c3279e47a59c80de60fa51ee98 : SUCCESS in 22m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6f0ee924dd84013b7ee3a2b4d97a78f : SUCCESS in 1h 03m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5f4c34ba7ef64a16a4535e42a5713a23 : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bb34c45612394bcb9b466ee9292f1cc2 : SUCCESS in 1h 07m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9cb59a2c6191479fad08683e15be66ba : FAILURE in 1h 01m 25s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a3b79932379f48a281ffa8dc3ce46733 : SUCCESS in 28m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bcc3765386641d3a571313463347a93 : SUCCESS in 36m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4bb2cc444830491a8c22a78a914c0af0 : SUCCESS in 36m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e70654e074cf4782b988d10fb8305ef1 : SUCCESS in 49m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d584b9baeca843348825058ac7b0eb8f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3bf84dcf187407a9e736997036585b0 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/534a400852ee468e85533c356b1d7bc2 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aad4e925a924ef6bef10c046767a2f3 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302f75c5c0a346d3a469613619b875fb : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb3b7d455e534458a6027f49ac15f314 : SUCCESS in 20m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85bc4d9e827243c8ae41c7ec9c2b706b : SUCCESS in 18m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a691dc93dff42cfbdb4a64f8237768b : SUCCESS in 54m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/46f83e509e4f476b803248fa6564cc23 : FAILURE in 49m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d9fc3c5af0e49a08277427aac671a60 : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":3},{"id":"14069c9cfeb192f6b692d4c9ee2cd20859e67bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:51:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3be97068d5474a5680f0060112d57c34 : FAILURE in 34m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/513c4b7e37c44939ae176ca87b95f4ce : SUCCESS in 33m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d3be3d8419143589ff9c9e2a615e3c7 : SUCCESS in 1h 21m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c9c94059abe4ab981624e2af53f800f : SUCCESS in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4ede93da5c73d69854f5fa3713c5b712fda2acd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-02 16:51:50.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"459f5d685bc8e1c46e08b1e14a0ee51aa0912086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:52:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f6f78be9deef98dd9a1f7a8df5d48e10a24f5d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:13:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/aabdd7053a544f3996fd1189949e22f1 : SUCCESS in 28m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eb845cd446a5483aaa1fba9fcfc97a29 : SUCCESS in 1h 09m 23s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/83802652142842a8949001e480fe771e : SUCCESS in 1h 13m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/772dbfff5a7e45ed8f784bde01def29d : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a814029007547d6a886993fc4370f84 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2b93c244042486da3d093ff5a7b1a99 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbed32577b774885a12ec646464dfa76 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2027faefe4934550a4317f2c8a56c3ff : SUCCESS in 9m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6b3a5c960d14728860dbbeedbd08a24 : SUCCESS in 22m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d327750bfd934ac396642a6d6b4b69eb : SUCCESS in 56m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9ed4e928d544d379aef9f12dbccccd7 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b57e8800cb5945979a0b2cf4ca3f2e907702880c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:13:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3d79ed653cb34618db8d5e289f636a09277b0dc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:15:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20b5001d3f8417a9ce51ebd96547f3d : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"23631a236935027e13c70633078d243a196bfc5e","revisions":{"87086fd876de524c0c0e368fdb65c4436663a650":{"kind":"REWORK","_number":1,"created":"2020-09-02 09:14:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/69/749469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/749469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/749469/1"}}},"commit":{"parents":[{"commit":"12f1f7e4de275d98dd136da02ac5676244d9c752","subject":"Merge \"Clean up the stale FIXED message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12f1f7e4de275d98dd136da02ac5676244d9c752"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 09:11:12.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 09:14:32.000000000","tz":120},"subject":"Work around setuptools\u003e50","message":"Work around setuptools\u003e50\n\nsetuptools 50 introduced following failures during pip install:\nModuleNotFoundError: No module named \u0027setuptools._distutils\u0027\n\nChange-Id: Idf401536b482b4088f5fb66408917752fd71726c\nCloses-Bug: #1893920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87086fd876de524c0c0e368fdb65c4436663a650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87086fd876de524c0c0e368fdb65c4436663a650"}]},"branch":"refs/heads/master"},"175f6cda1661f5ce305c2fde56ffc48bdcbf14e2":{"kind":"REWORK","_number":2,"created":"2020-09-02 10:20:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/69/749469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/749469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/749469/2"}}},"commit":{"parents":[{"commit":"12f1f7e4de275d98dd136da02ac5676244d9c752","subject":"Merge \"Clean up the stale FIXED message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12f1f7e4de275d98dd136da02ac5676244d9c752"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 09:11:12.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 10:19:52.000000000","tz":120},"subject":"Work around setuptools\u003e50 and fix bifrost network config","message":"Work around setuptools\u003e50 and fix bifrost network config\n\nsetuptools 50 introduced following failures during pip install:\nModuleNotFoundError: No module named \u0027setuptools._distutils\u0027\n\nSet bifrost internal_ip variable to adapt to change [1].\n\n[1]: https://review.opendev.org/#/c/748230/\n\nChange-Id: Idf401536b482b4088f5fb66408917752fd71726c\nCloses-Bug: #1893920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/175f6cda1661f5ce305c2fde56ffc48bdcbf14e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/175f6cda1661f5ce305c2fde56ffc48bdcbf14e2"}]},"branch":"refs/heads/master"},"23631a236935027e13c70633078d243a196bfc5e":{"kind":"REWORK","_number":3,"created":"2020-09-02 15:23:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/69/749469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/749469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/749469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/749469/3"}}},"commit":{"parents":[{"commit":"12f1f7e4de275d98dd136da02ac5676244d9c752","subject":"Merge \"Clean up the stale FIXED message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12f1f7e4de275d98dd136da02ac5676244d9c752"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 09:11:12.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-02 15:23:13.000000000","tz":0},"subject":"Fix bifrost network config","message":"Fix bifrost network config\n\nSet bifrost internal_ip variable to adapt to change [1].\n\n[1]: https://review.opendev.org/#/c/748230/\n\nChange-Id: Idf401536b482b4088f5fb66408917752fd71726c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23631a236935027e13c70633078d243a196bfc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23631a236935027e13c70633078d243a196bfc5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
