)]}'
{"id":"openstack%2Fkolla~749655","triplet_id":"openstack%2Fkolla~master~I3d9547237ffa3ee02c306376f0e553fce5472e57","project":"openstack/kolla","branch":"master","topic":"cradox","hashtags":[],"change_id":"I3d9547237ffa3ee02c306376f0e553fce5472e57","subject":"Replace cradox with rados","status":"MERGED","created":"2020-09-03 06:56:35.000000000","updated":"2020-09-07 09:20:29.000000000","submitted":"2020-09-07 08:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749655-1599469030317-69a35627","meta_rev_id":"8299b5e81c82ce2f513e39536e85e560a0af66bf","_number":749655,"virtual_id_number":749655,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 08:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2020-09-07 08:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-09-07 07:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-09-06 09:04:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-06 09:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2020-09-07 07:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-07 08:56:42.000000000","_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-06 09:04:14.000000000","_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 09:23:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-03 13:39:02.000000000","updated_by":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"reviewer":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"state":"REVIEWER"},{"updated":"2020-09-07 07:41:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-07 08:56:42.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-09-07 08:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d33a76f5a20378b8223a3e3c26f9a02e076e489","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-03 06:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e58beeb8217f732618ae1f947f633f45d303266","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 08:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/88e8c8fa9e0a4f7d81cc1f36bed0ff4f : SUCCESS in 43m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8a40bc6e63d4195b38a133a85e75e76 : SUCCESS in 35m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/96fc422efd594f9c927b8b90e5c85d51 : FAILURE in 58m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/660dea4a4f934cdd8e858652b2c09433 : SUCCESS in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e06c773236ab93971762e4f4c3bdcb77c601e312","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 08:12:27.000000000","message":"Patch Set 1:\n\ndo users have to change anything?","accounts_in_message":[],"_revision_number":1},{"id":"dff673ea09d192b4ade7f378dcafd55f5bd3c95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 08:17:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2a34d09b23b145029c2d926e88ba2c4f : SUCCESS in 21m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8f8849f91549480b9aac1ea7052292af : SUCCESS in 21m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3235fa40b70b4632a53b51bd5b02ea33 : FAILURE in 45m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7c488808b3be48dda4266ac209d59619 : FAILURE in 44m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9593313ae5174577af2bca5485c09673 : FAILURE in 1h 06m 58s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d34489bc24024e049f09b241b40be850 : FAILURE in 1h 13m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ebd710cac2234dd9b646529cab076c9e : SUCCESS in 27m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d66aa86f62b844c8bb2bb8b2744a7e0c : SUCCESS in 20m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a5ef7b89832942e8ada6276e7427d2f9 : FAILURE in 31m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dfc632a5d084c9db9cebf5dda631a24 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0ef57c443e41d9a4d7e9fb179b0ace : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d22dc3ad64f4725a0d76f82ac3373df : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec3d93cd25b64141a550b7a0c467cf63 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3329e05f8745aaa96b86733052edda : SUCCESS in 7m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18ae7f20fedf4ea4b0bd8ccf06ce07f9 : SUCCESS in 19m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/560a702e23884bc79685c5b21396aacf : SUCCESS in 19m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9453be1904964059a3f7b21f326ca85a : FAILURE in 37m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7e88e8ea88104026813cd8647b0c40c1 : FAILURE in 39m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0098954ba18c45cba3f4de60f9d367f3 : FAILURE in 1h 00m 11s","accounts_in_message":[],"_revision_number":1},{"id":"85c4741a60363947efef52b3562079c07c3fc8f1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-03 09:22:26.000000000","message":"Patch Set 1:\n\nthey don\u0027t. cradox has been merged into ceph a while ago.","accounts_in_message":[],"_revision_number":1},{"id":"4e47c98b2f28633b9a6375c71cf7938c62683c4a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 09:23:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97973b5681b2f5c0bf6d4c547afdfa34a0853edc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-03 09:23:42.000000000","message":"Patch Set 1:\n\nbuild failure due to glance image can not find boto3.","accounts_in_message":[],"_revision_number":1},{"id":"ca09c5353c56449899ff641694569122dab9b026","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 09:23:58.000000000","message":"Patch Set 1:\n\ngate blocked by https://bugs.launchpad.net/bgpvpn/+bug/1894056","accounts_in_message":[],"_revision_number":1},{"id":"6d5211b573d4af992b3a74a82696ddd4832ae040","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 09:24:42.000000000","message":"Patch Set 1:\n\n\u003e build failure due to glance image can not find boto3.\n\nno, that\u0027s just the best false alarm we currently have :-)","accounts_in_message":[],"_revision_number":1},{"id":"453781bf5d27ee152b734c950aa79d90b7a24c33","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-03 11:17:22.000000000","message":"Patch Set 1:\n\nfrom a different failing job (this time on ubuntu):\n2020-09-03 07:30:59.922 6 ERROR stevedore.extension [-] Could not load \u0027glance.store.s3.Store\u0027: No module named \u0027boto3\u0027: ModuleNotFoundError: No module named \u0027boto3\u0027\n2020-09-03 07:30:59.938 6 ERROR stevedore.extension [-] Could not load \u0027s3\u0027: No module named \u0027boto3\u0027: ModuleNotFoundError: No module named \u0027boto3\u0027\n\nAt least, the error message is consistent","accounts_in_message":[],"_revision_number":1},{"id":"b068f541e338537449abfee0d002b9009e93f66b","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2020-09-03 13:39:02.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2d271bf6a7c33ac544761e0e280571c2642d29f1","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2020-09-03 13:39:09.000000000","message":"Removed Backport-Candidate+1 by Pradeep Kilambi \u003cpkilambi@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"1581a5a4fd590d722ac01bfb49b98217b6537ba5","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-04 07:49:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f862e2c56248a27c91a543d341132d554feedbc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1d2e076ad8a488ba048d906f96e90a1 : FAILURE in 38m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81e1cc06efb44c41beec59a6d74bcd26 : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aeb2bdd1c5544999bf6ca9a4767353e6 : FAILURE in 48m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c759d00e77214b518b1a9f0074e8630c : FAILURE in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e61414e38ece99b5e4d2539597bf780c99d4b0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:05:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d75996023743498d9932ee977a856bf5 : SUCCESS in 39m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a5be392f3c704e8ca5e2897a9a2298de : FAILURE in 48m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eb7f326d3690444bbfb6dea3d78b7a08 : FAILURE in 52m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0cf5589f8594850834bf2cf2c9be1b9 : FAILURE in 42m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bce58db96e9749c78be9ca20a2e8b38c : FAILURE in 1h 09m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/721edcf91b83413ea9cc2b4476dd4ace : FAILURE in 21m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2be227178d534a9c909dcfbcbf58f806 : SUCCESS in 26m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2202002744d7478ea0f00d14f8147921 : SUCCESS in 32m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/232637d01f724c628ebdc417c0df0494 : FAILURE in 30m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4582d13059734d68866e3a0ea397eea0 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc38291efc4496185a661992d795245 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/777442da063648debbc16fd15ed66b21 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ea62327b1f4719a8d33a1411395c4b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bafb7d48e64e9bb94dfd6f884811d8 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bba0c1eee9f84b16863fef6ee954908e : SUCCESS in 21m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/169d46a60d934e468faed24a545287c8 : FAILURE in 25m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6098dea291c64d9485e284e7287c3914 : FAILURE in 36m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f2c08b8268f44c0994dd7263db3324f : FAILURE in 40m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56e090cb4eef48029f4e388013bf3ffd : FAILURE in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5b45dc9578a9907b63f15094916d245f1ab595f3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-05 16:16:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93c68e57d25e245f82d6deb0e3d78ba734eb4703","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 17:12:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0cb7f90314c48ef9d69335857866481 : FAILURE in 43m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f300e89e31524b1983a970674d7bb850 : SUCCESS in 32m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3d04cbce92140dd9af507c1b6b35fd3 : FAILURE in 50m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a2d79e22dc5d4d648faeb222c343bca4 : FAILURE in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b54e1c263a3473c8876265f23ba6ea5ed706df0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 17:25:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/520534db2ff34b1fbb6a932d549d90c2 : SUCCESS in 29m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d65ee9358e0647b8a637e6779993c1f2 : FAILURE in 32m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d525c123c1df47518bd587984325501a : FAILURE in 47m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/11c57562d21c485aa285553cce4e0b20 : FAILURE in 31m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1541dd80b9254d04a5d0beed1c9bb52a : FAILURE in 1h 08m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f327cab099048c180b19f95273d4f4a : SUCCESS in 25m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/147dbf24b2bc4c99b326e6bd33d1dc9d : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2222dc80530c4271929efcc1fb4febe9 : FAILURE in 43m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48d6cc3e0c30482cb8c4f0a2c6ff1806 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510bec11650f430bb3b09eb2c42812a5 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3f5c27703084f1781ab63d9822fe49d : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/173538f251364791bde8be7b1ee98469 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd31ff8cac8048b6832c6b3e4c10b5af : SUCCESS in 7m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0fc58ddc9514496aaf4239143ffe4a82 : SUCCESS in 16m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdf5d965d77e4a23a0b852a241dc7c0f : FAILURE in 28m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e719dddc75794f498d88c70d6d81fb7f : FAILURE in 37m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/155e9f676dbd4980bd55fd2b4f0c75c5 : FAILURE in 27m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/449d330a876a46dfa79c8e55b685674d : FAILURE in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bc61a59cba72e091beab2aa259be984221ddc0fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-06 09:04:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e8281be1ff8cd2e4d89a99c783be310024bb9ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 10:36:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/66a408701ea34d89bd7dce9edec084dc : SUCCESS in 31m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/25151ba055cd46489bb3bdfd4470c389 : SUCCESS in 27m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/52870c25317748eca691a798ca315216 : SUCCESS in 1h 31m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2413952988fc4591a711539fa7dfa98a : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0545bc425e3e4aeda830098ece6cb5f0 : SUCCESS in 1h 08m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc39e85260f747c7bd8160a245e8302b : SUCCESS in 16m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/38eff79a99564f53ad1e1b89cf62605a : SUCCESS in 14m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab23b6af775647808aa086d038c5a4b5 : SUCCESS in 52m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bb4f7e8ca7243c580d048b085b7877b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570c6dc1cdbb48ce8c7a5f02a9666bc1 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/762769560e2941018b57bfb9267e3e54 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d719ffc7922e4c49b70627abc65b05f5 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed35f0506a744902817797fc5476d05a : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2094135c7c5a41319e2b42ba9626c75b : SUCCESS in 59m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1771e16fa678454bb6b3cc0e0a9afdd8 : SUCCESS in 16m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/775abf56e4a84a99a504b39a11b280c2 : SUCCESS in 1h 22m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5c4ca11d1bf40bcaf53fbca3a8c3453 : FAILURE in 35m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf4af0ed7dac4fb5b3bde1c061525f9b : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7efa48dbbc6fb2ec0d45ff671e28c2db0ea9753c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 11:04:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/34a65f103d664da3a76b6b8ac321d207 : SUCCESS in 34m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72d01e32f778463495f5d81c7629982e : SUCCESS in 1h 15m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a461a04c06314dce97b8e85efcc8e26a : SUCCESS in 1h 51m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/45e9d6d1e4dc4d2b9d2859d6be2bba1f : SUCCESS in 1h 06m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ca098a4a0869222585518c9f798958bf310b823","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-07 07:41:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7ad2a276f3fc3b2750470a12fcf4640ffedf4142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:41:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cf9c6c55e60c28d7a2e284cc11caef249d82ef5","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-07 08:56:42.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"511bd204a2c827372a12d572e0a6bf2384c25e9f","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-07 08:56:54.000000000","message":"Removed Backport-Candidate+1 by wu.shiming \u003cwushiming@yovole.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"d10322e54be685750bbe2075d4534363b545df44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:57:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e78f6d02883a49f29154f4bcb736bbfd : SUCCESS in 18m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f8be90545e4140d1bfaba1a521ba4763 : SUCCESS in 1h 04m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6b07f18223ad485f8cecb0060adda29b : SUCCESS in 1h 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57b3d7f8dcc94aa0a42d03474dd5874b : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e404c1bc74f948f9804603baf7f3df1d : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba7ee22e4f5842dfa4e617d281dd7269 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e501713b82749d994e03558b3e12103 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921ade602e094627817bbab8c39d9d92 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/934d10a528cd47aa914f6ad110bd37c9 : SUCCESS in 18m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec20f135d0b64380a1fff4e0dd068057 : SUCCESS in 54m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e90ff1d4f93428eadcb5b25a0108c58 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":2},{"id":"450a097d521bb4efd81e68b93b6e11c431c0fb21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:57:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e9c9cd4af1e1e7babcf82d94bf6a2f6056d09d4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5980fde21575401cb3902b7e88fff420 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2},{"id":"d16f67b3795139de19669455ae75e6e7df877a8d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-07 09:20:29.000000000","message":"Patch Set 2:\n\nThank you.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"934d0b21f8f709f44f795b6629ddf8cfd65f7be8","revisions":{"6195b3d9ed53b748cad5a138947a3b3a0475b6c5":{"kind":"REWORK","_number":1,"created":"2020-09-03 06:56:35.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/55/749655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/749655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/749655/1"}}},"commit":{"parents":[{"commit":"23631a236935027e13c70633078d243a196bfc5e","subject":"Fix bifrost network config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23631a236935027e13c70633078d243a196bfc5e"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-03 06:55:48.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-03 06:55:48.000000000","tz":120},"subject":"Replace cradox with rados","message":"Replace cradox with rados\n\nGnocchi supports both, but cradox is not actively maintained.\n\nChange-Id: I3d9547237ffa3ee02c306376f0e553fce5472e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6195b3d9ed53b748cad5a138947a3b3a0475b6c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6195b3d9ed53b748cad5a138947a3b3a0475b6c5"}]},"branch":"refs/heads/master"},"934d0b21f8f709f44f795b6629ddf8cfd65f7be8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-06 09:04:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/749655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/749655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/749655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/749655/2"}}},"commit":{"parents":[{"commit":"2725ed9d842622078e31165d91718e58d07620e0","subject":"Emergency fixes to pass CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2725ed9d842622078e31165d91718e58d07620e0"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-03 06:55:48.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-06 09:04:14.000000000","tz":0},"subject":"Replace cradox with rados","message":"Replace cradox with rados\n\nGnocchi supports both, but cradox is not actively maintained.\n\nChange-Id: I3d9547237ffa3ee02c306376f0e553fce5472e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/934d0b21f8f709f44f795b6629ddf8cfd65f7be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/934d0b21f8f709f44f795b6629ddf8cfd65f7be8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
