)]}'
{"id":"openstack%2Fkolla~749881","triplet_id":"openstack%2Fkolla~stable%2Frocky~I79ee884ad35bf5166c9ea303d73416271ff5f637","project":"openstack/kolla","branch":"stable/rocky","topic":"fix-ironics","hashtags":[],"change_id":"I79ee884ad35bf5166c9ea303d73416271ff5f637","subject":"Fix ironic build issues - LANG related","status":"MERGED","created":"2020-09-04 07:22:20.000000000","updated":"2020-09-04 11:06:37.000000000","submitted":"2020-09-04 11:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749881-1599217597175-5b9a2423","meta_rev_id":"34dfc8dec958e148c63acdffd4cd905be8f1ab2f","_number":749881,"virtual_id_number":749881,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 11:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-09-04 07:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-04 09:33:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-09-04 09:33:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-09-04 07:22:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 07:22:31.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-04 07:48:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-04 11:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8a7aa6ffb1a4bf11cc9a34866b76333dbd6d7f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e90d3d6ac3719f851ed272c1f830ba1aa64abdbe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-04 07:48:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cb35c57c7377ad839dcdd18f0ebf8f302c0e5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4213866dc07e4e62b83ea5250fc41220 : SUCCESS in 51m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/83989be4bfa44e31a8d4a5156a59ecac : SUCCESS in 54m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/58c902e41c3e4d8abac7258729e155fe : SUCCESS in 1h 20m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0be02248d8e5495ea2d9736170bdecf2 : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ea41fa41d3734335bf0d17db48c224dd : SUCCESS in 59m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/60def570718047a0a01667309f5e3c88 : SUCCESS in 43m 01s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/940549a3bb1f4806b5fa615b2020ded9 : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1e4ad571aca34d1daafe94278ea9ef7e : SUCCESS in 1h 04m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f711e594a47f4268b6a212637143df2a : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6cc07caf3647a89d94bbcd9eb0c4c3 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b1c0a9c05d84548a7266e7723fe003c : SUCCESS in 7m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bebdb68859b147e0aee30b7269dcb80c : SUCCESS in 7m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9acead6efc2f4ad7826504252f3cc82c : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07133f5aae014f7298039d1e2e6cabee : SUCCESS in 1h 23m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70816dfd90e346b3992ea5dff6178216 : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/295d05d0ce89430ca4386eb0460baa1d : SUCCESS in 1h 12m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/688e98cdcf384e60a789a857c1f438f0 : SUCCESS in 1h 27m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"a1363997f323e1b357ccf94b8a746c15ef69460a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 09:33:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca0cbec5e055af060f9e89c418f5b00091a05f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"468aa88750b851ed3727ebeac991fb090a8830fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe3e5b5265d24f5786909d41d5c06e8a : SUCCESS in 34m 55s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/60a47de4c8894cd1887303e88262099f : SUCCESS in 46m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06feff3a602f4a8ea55819a8cf91074b : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf7d71c259d4b77b2ee8c906b10e386 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f3751f992aa4791b7d2dce3c4ff6447 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/031a2cb0ddc3478cb8a24939b5a9022a : SUCCESS in 4m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7fa84df2095043edb4e6c0a70e9f99af : SUCCESS in 4m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4b74613b4f64edd959965097e835816 : SUCCESS in 26m 09s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d043ee4a951c44ce6acfdce6a1c32140ad4c5811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"965a288da2e3c0f170167279990ddeeb5f1de054","revisions":{"965a288da2e3c0f170167279990ddeeb5f1de054":{"kind":"REWORK","_number":1,"created":"2020-09-04 07:22:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/81/749881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/749881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/749881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/749881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/749881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/749881/1"}}},"commit":{"parents":[{"commit":"5b64da9f319162c9d17e2894d8f5b0658d661e75","subject":"Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b64da9f319162c9d17e2894d8f5b0658d661e75"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-28 10:20:02.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-04 07:22:09.000000000","tz":120},"subject":"Fix ironic build issues - LANG related","message":"Fix ironic build issues - LANG related\n\nironic drivers-requirements installation fails with:\nUnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc5 in position 2:\nordinal not in range(128)\n\nSquashed ,,cherry-pick\u0027\u0027 from stable/train - adding LANG block and locale-gen\nfor Ubuntu family in base Dockerfile:\n\nhttps://review.opendev.org/#/c/684808/\nhttps://review.opendev.org/#/c/691835/\nhttps://review.opendev.org/#/c/694099/\n\nTemporarily blocks bifrost building as well - to revert!\n\nChange-Id: I79ee884ad35bf5166c9ea303d73416271ff5f637\n(cherry picked from commit 3c435cb0781b343ad7f7223609cf2eefb2f8738a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/965a288da2e3c0f170167279990ddeeb5f1de054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/965a288da2e3c0f170167279990ddeeb5f1de054"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
