)]}'
{"id":"openstack%2Fkolla~750136","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I3d9547237ffa3ee02c306376f0e553fce5472e57","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I3d9547237ffa3ee02c306376f0e553fce5472e57","subject":"Replace cradox with rados","status":"MERGED","created":"2020-09-07 08:24:40.000000000","updated":"2020-09-08 09:38:53.000000000","submitted":"2020-09-08 09:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750136-1599557837213-c09e01d6","meta_rev_id":"92d331365c5cbb19e77757a109ea1728affb3b91","_number":750136,"virtual_id_number":750136,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 09:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-09-08 07:45:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-07 11:30:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-08 08:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-09-08 07:45:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 11:30:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-08 07:45:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-08 08:46:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-08 09:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bef74ce2218cb8a2d7bc416d1ed19b382e740219","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-07 08:24:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"650d3cfe4a3461defeca60cceb6a15b4c971578a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 09:02:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8fa102a4a221d0036b02b853dcea74af73acef60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 09:03:27.000000000","message":"Patch Set 1: Code-Review-1\n\nINFO:kolla.common.utils.gnocchi-base:No match for argument: python3-mod_wsgipython3-rados","accounts_in_message":[],"_revision_number":1},{"id":"b30809769d4ce986cf3e1747721e7d954d5166c1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-07 09:18:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3cb5b6de3ec5d5a96099a8cdbec85c35b9816bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:49:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/18a5111d3cbb47a996552cb23822bf0d : SUCCESS in 58m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/235f987defbd4a14be31edb95f30c259 : SUCCESS in 50m 28s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8c0bd4b9de874a378db5799110c88cbc : SUCCESS in 39m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d1ab2da9e654fc08652ca4a612f74fd : SUCCESS in 49m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ea3736d91bb24303915d3a53edf146d3 : SUCCESS in 1h 21m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/792e928d615440bbb8c5d95bd6f0c0fc : SUCCESS in 1h 11m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/36477a7cf1554b61ab4940dad2ce56a6 : SUCCESS in 1h 20m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/61d13b9b35144b6494ae59f51cd1436c : SUCCESS in 1h 12m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ee1c82e555e845819163c3e88ff8cf97 : SUCCESS in 1h 29m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a011dcec68494957b58b4b31a1a20fd1 : SUCCESS in 53m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7eacff274f514af688e514b084732d6f : SUCCESS in 31m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c9ceec8aaa8487f8c0d684900bdc057 : SUCCESS in 30m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f6b7e1eadb041b797823a20ef2f4b0c : SUCCESS in 1h 01m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/050242fb4dfe4bc88526017e15d8b0a7 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117aa7545dc94ea7a57542fa114b3707 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb7a40b60e144145a26b221d3a896481 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/993ee30fd7cd4020ad6bb381b8b3c445 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a87bc5cface4acb847f2a68aea27724 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8ca90569e442aaabcbd14f8a24b934 : SUCCESS in 5m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42c2e0c3f2de4509ab5d107cdda475df : SUCCESS in 37m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0af80fa187946e1919fcf1a4a623ac9 : SUCCESS in 25m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c481f7a6ae34b2c901a137529319dc6 : SUCCESS in 54m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6aa33fc9ddb47b990af69530d64ddf1 : SUCCESS in 54m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e7f72697a02456ab56df1deccfc36bd : SUCCESS in 59m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"04876cdc16df433edb047bf6275dd9a2532d07d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 11:30:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3620b21ff5aae5ab4be94c5873566429abcc7fdd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-09-08 07:45:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"10bb8ad65dea6cc2c45a5be11f1e273ef50b6cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 07:46:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c69d5ec569afd9cd22cf45bbbddb27d17d9dd45","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-08 08:46:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7e71333057cae616b9d9cc7d326d272f5be1eeaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:37:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/60caedc1f0c74561883a6e3595b24424 : SUCCESS in 48m 01s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/99edb6cfc0c64637816a49ce6d0669e9 : SUCCESS in 24m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd9ced7534de498eadca69b3aef1ccb7 : SUCCESS in 1h 27m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e01cecc632b5401bbd9ef7cfdf6a3b1b : SUCCESS in 43m 54s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a2bea128c650478fb8a6227b67468577 : SUCCESS in 1h 42m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f2a6706f310d413b9b4bc5a944c94103 : SUCCESS in 35m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47968495d60040ef81cd9ce37efc3d01 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5113e4d5524d4a77af6830826dfc1cac : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6988a6b44fe24bf5a14ea88c2be829ac : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d9a2848eaf840bcb629abe34f9a9eb3 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba33020a31944745b7bd1866581a1401 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455c7c1ef0a346b38d58d626f88b4bb6 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81f377547eac4969a7269109bb89fa28 : SUCCESS in 21m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdadd467fb5c467f86738279438affba : SUCCESS in 36m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4cf6176f9c744c9986647dff9fe075fb : SUCCESS in 1h 02m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"ab0ba934f4edf8011790538661f8b31eb7f12e68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:37:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"92d331365c5cbb19e77757a109ea1728affb3b91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:38:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a401c7db60e4eb49cb29f5f0f4a8c39 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6","revisions":{"06acc7a033c65fef38a821d177664acc36587387":{"kind":"REWORK","_number":1,"created":"2020-09-07 08:24:40.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/750136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/750136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/750136/1"}}},"commit":{"parents":[{"commit":"cd4d0ae1f90665424ebd925f4f6dbcbc526709a2","subject":"Merge \"Use upstream Erlang on CentOS 8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d0ae1f90665424ebd925f4f6dbcbc526709a2"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-03 06:55:48.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-07 08:24:25.000000000","tz":120},"subject":"Replace cradox with rados","message":"Replace cradox with rados\n\nGnocchi supports both, but cradox is not actively maintained.\n\nChange-Id: I3d9547237ffa3ee02c306376f0e553fce5472e57\n(cherry picked from commit 934d0b21f8f709f44f795b6629ddf8cfd65f7be8)\n(cherry picked from commit 9bbc7531e249032ad6f5647863db21d88322f7f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06acc7a033c65fef38a821d177664acc36587387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06acc7a033c65fef38a821d177664acc36587387"}]},"branch":"refs/heads/stable/train"},"5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6":{"kind":"REWORK","_number":2,"created":"2020-09-07 09:18:19.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/750136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/750136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/750136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/750136/2"}}},"commit":{"parents":[{"commit":"cd4d0ae1f90665424ebd925f4f6dbcbc526709a2","subject":"Merge \"Use upstream Erlang on CentOS 8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d0ae1f90665424ebd925f4f6dbcbc526709a2"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-03 06:55:48.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-07 09:18:04.000000000","tz":120},"subject":"Replace cradox with rados","message":"Replace cradox with rados\n\nGnocchi supports both, but cradox is not actively maintained.\n\nChange-Id: I3d9547237ffa3ee02c306376f0e553fce5472e57\n(cherry picked from commit 934d0b21f8f709f44f795b6629ddf8cfd65f7be8)\n(cherry picked from commit 9bbc7531e249032ad6f5647863db21d88322f7f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
