)]}'
{"id":"openstack%2Fkolla~750154","triplet_id":"openstack%2Fkolla~master~Ic96032b8f0912b36f85824cffdd48f944669604a","project":"openstack/kolla","branch":"master","topic":"depends-on-bifrost","hashtags":[],"change_id":"Ic96032b8f0912b36f85824cffdd48f944669604a","subject":"[WIP] Test bifrost for failure","status":"ABANDONED","created":"2020-09-07 11:52:32.000000000","updated":"2020-09-08 17:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1cb498048890804dbec541567c964f7f040a6a09","_number":750154,"virtual_id_number":750154,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-08 12:34:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-08 11:57:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 11:57:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acb22c630ce70d929449975c31d0c013e261340f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 11:52:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b478d40b79139e91d2310ab9d8455275d304a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:03:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/76f47796b32a471ca26e79969ca0c815 : SUCCESS in 22m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3a86e37e198b4c719ca4018837da4b13 : FAILURE in 26m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/639a8ff3dffb47a2810665c6e31773d2 : SUCCESS in 46m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ed7714f28f814e85af3e09f0d804d169 : SUCCESS in 49m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3513500be770456fab6be383a66c6332 : SUCCESS in 54m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af478ce7ee504ed4a90b2b1a1cb2c346 : SUCCESS in 13m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8333d3f38d40409da784744bc62fd859 : SUCCESS in 18m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ea87c20f9f6a46cc8cf375ad52e7ef66 : SUCCESS in 43m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f42833deb88a486ab0cc3d39edd2c3dc : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abab028f884447fa892c2161d9feca82 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc1bd96710c24128bcb67f9da4d2ed32 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8a25ab7c98946b0bdf6942a998a12b0 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b4981afd2f4f038ab7fc727a8d6d42 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a163bc3abfa742f48c6b9e8a0af3997d : SUCCESS in 22m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a21c59de16e64069bc738084e64a035c : FAILURE in 17m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e58bb212ae948ce8c30c1e4d12f9b93 : SUCCESS in 44m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd5503d03c80461a9784a313616fa975 : FAILURE in 31m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67a3a60ac28447f6b06916c93224ca5b : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"db7f34a0ac6cdbac748fa831a53ce9158cfac81b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4af8253aef644fd9eaf573690e7ea6f : SUCCESS in 43m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdf0275eb22344d9a64cc613c83ad177 : SUCCESS in 26m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/059700b7539846a9872a068f8a38fc2d : SUCCESS in 1h 18m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a242cf0339d7419f90d404dcfd043373 : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc08da182d67e75676d5e20999e18b29b0922637","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 14:42:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"36a2ab7606e3438ffa6bc6e09e8234da3a9fcd42","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 15:38:01.000000000","message":"Patch Set 2:\n\nok, dep merged so zuul ignored it","accounts_in_message":[],"_revision_number":2},{"id":"f67adf75a9ac84c330cd25fc3041793c2c7949f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 15:38:13.000000000","message":"Patch Set 2:\n\n(no bifrost in projects)","accounts_in_message":[],"_revision_number":2},{"id":"d04c819284b5ff6670e26dafa938c9f394dd3c30","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 15:42:39.000000000","message":"Patch Set 2:\n\notherwise bifrost passes now, let\u0027s test breaking","accounts_in_message":[],"_revision_number":2},{"id":"5a86e133be65d4e635d72ca8a8be892812f37603","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 15:43:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b51226030e4a67af53385b4c885c34b4fb18e031","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-07 16:51:14.000000000","message":"Patch Set 3:\n\nit works (breaks) as desired! :-)","accounts_in_message":[],"_revision_number":3},{"id":"86f1db2fb4d6178a3357fdacdd005c63de7e9a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 16:58:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7f7bd2999c8b4d56abb53fcb327dd79e : SUCCESS in 23m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/79c16cd1ccbe4329bac28200383dacf4 : FAILURE in 27m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1a25a392729c4198940d78b2a70e7543 : SUCCESS in 1h 05m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9027731045d040ee9a37e421d0de3a07 : SUCCESS in 59m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/86b75ed3720e43da80b9985aabfe7f1a : SUCCESS in 1h 08m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb85973021e24f9fb2b307aec7b66802 : SUCCESS in 16m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/253dc60a46de4a4482c9b7f4089a79bf : SUCCESS in 17m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5db1968b2ae744999dd7c770dade7a80 : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c35e6e93c02a48c9b2e3194601e42e28 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a9c83a24c74425896d846f9c545193f : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7646e0dbfc6940f59e9debef191d5010 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fd972c3a0cb4cb6ab9cd2ba97a74866 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef7fdfe6e7647719a0beba034366f11 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3c97f06dbc44053ac249d7d16964bd3 : SUCCESS in 17m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ee0cf243b23497f9bf340a4b041a5de : FAILURE in 21m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfccd9cf090c4a3f9a4d54098522b315 : SUCCESS in 48m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/89a60b234c2748d986b7637d36c6b4dd : FAILURE in 40m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52064ecf5a0a47f59411686529692b53 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2980014fcf78c362b97aab4d590a7e8fa57df1a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 17:03:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b60f4a98c964fc4aa7667a6807e29b8 : FAILURE in 38m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9455df6bef94f419a42489b7cc70a9b : SUCCESS in 26m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2be09e7fadba4bf7954f11dadac88c36 : SUCCESS in 1h 17m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/99c6923e1fc3431facb53bb134b502b6 : FAILURE in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ead40518913cfb15e79a454481a77cb49b4ac0a1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 09:46:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f67d8e7a0516f7f4612c3757fe244f30865bbfaa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 10:15:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"479a0071493a130c53b4cbc5625ae24b0f3ab40c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 10:40:17.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3a33555f5e1d47f278a39c83fdd0443372d0de92","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 10:42:22.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"c46cd9c0594cef954523e82e6ac124b5b99dcc43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 11:57:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e2497da4ad7346abbdfa17336871e075 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bf3e42a6b773459ba56fa7ea7de2bf12 : FAILURE in 25m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d15232f0de4a45f4a71171a3280e9531 : SUCCESS in 1h 02m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d12bc8f6cd4346b19dc2221c3f51797b : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/702a8e534a4f4fa29de79a0b3d478f2d : SUCCESS in 1h 09m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6c722e7d76bb43dcb935eb99e4a5350f : SUCCESS in 15m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/314e5cad60304b8fad7d87d1078f72ec : SUCCESS in 17m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0511d22f7735456f998cf4af0b3cc029 : SUCCESS in 55m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e54af053d624aba95f73c84303ae1ba : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d41a955f8a40c58c1252b63a7a288c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1602336742dc4995abf2eabdab84ecb5 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd5b521696447ffb2aac41f4ba72f14 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32cdd829e7540a3b1efcca2b36f8bfd : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d59d4c7e3f049d7bfaee66234f2b93f : SUCCESS in 21m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c46fe36498c4f0cbc4e7c716965f3da : FAILURE in 25m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad235ae1b35140c7a0b5eb2e2b636504 : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c14969ba5e7b4e85ad076cca215152bc : FAILURE in 36m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa138b6bac5d4b63b1e4ff1dc72bfd1e : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":7},{"id":"3687adcab89193a65d47d1a3fb0a24933dbbb00c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:21:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a53af13505834340937f0ee21e410c59 : FAILURE in 54m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b00609c786d34597825f97e7e6a5cb03 : SUCCESS in 42m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6eec344713e4ecfa5bba75821c8e26f : SUCCESS in 1h 28m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/50be7ecf9b5d4abb80e24537aff82688 : FAILURE in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e1f5e40b9a20991430e8b864fe76974704d70e1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 12:34:31.000000000","message":"Patch Set 7:\n\nyeah, works","accounts_in_message":[],"_revision_number":7},{"id":"6f33a8307dff4cde2042a734f9723efb19218118","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-08 17:22:35.000000000","message":"Abandoned\n\nno need","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3728e7cbb3a4ae8f3a75231e281ffcda827b3731","revisions":{"29baf4258133fbac19888a96d10c56dae5876ada":{"kind":"REWORK","_number":1,"created":"2020-09-07 11:52:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/1"}}},"commit":{"parents":[{"commit":"9a6f8d3ce970d50560adb04887a3a866fcac1b58","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a6f8d3ce970d50560adb04887a3a866fcac1b58"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"subject":"[WIP] Restore bifrost","message":"[WIP] Restore bifrost\n\nDepends-On: https://review.opendev.org/750146\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29baf4258133fbac19888a96d10c56dae5876ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29baf4258133fbac19888a96d10c56dae5876ada"}]},"branch":"refs/heads/master"},"3cc7c06582fb3140d95a59472a7636951d902288":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-07 14:42:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/2"}}},"commit":{"parents":[{"commit":"1d2286cabfe68710e3df82b711e87810f3e3a401","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d2286cabfe68710e3df82b711e87810f3e3a401"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 14:42:55.000000000","tz":0},"subject":"[WIP] Restore bifrost","message":"[WIP] Restore bifrost\n\nDepends-On: https://review.opendev.org/750146\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cc7c06582fb3140d95a59472a7636951d902288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cc7c06582fb3140d95a59472a7636951d902288"}]},"branch":"refs/heads/master"},"35d27c733dfa4d3cc9ad0648ad6e943b8c8f226f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-07 15:43:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/3"}}},"commit":{"parents":[{"commit":"1d2286cabfe68710e3df82b711e87810f3e3a401","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d2286cabfe68710e3df82b711e87810f3e3a401"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 15:43:16.000000000","tz":0},"subject":"[WIP] Test bifrost for failure","message":"[WIP] Test bifrost for failure\n\nDepends-On: https://review.opendev.org/750195\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35d27c733dfa4d3cc9ad0648ad6e943b8c8f226f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35d27c733dfa4d3cc9ad0648ad6e943b8c8f226f"}]},"branch":"refs/heads/master"},"db9ad4851a22ebd3592cba17aeced380fb729bfa":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-08 09:46:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/4"}}},"commit":{"parents":[{"commit":"0ea545f6c0b013c4c11be968079c5f55dcbb745e","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ea545f6c0b013c4c11be968079c5f55dcbb745e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-08 09:46:09.000000000","tz":0},"subject":"[WIP] Test bifrost for failure","message":"[WIP] Test bifrost for failure\n\nDepends-On: https://review.opendev.org/750195\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9ad4851a22ebd3592cba17aeced380fb729bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9ad4851a22ebd3592cba17aeced380fb729bfa"}]},"branch":"refs/heads/master"},"3632512dcafa8adb520bbb4b354f6edc40691663":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-08 10:15:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/5"}}},"commit":{"parents":[{"commit":"1902310ad75aa969dfc7db99d92f072ae99637ca","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1902310ad75aa969dfc7db99d92f072ae99637ca"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-08 10:15:00.000000000","tz":0},"subject":"[WIP] Test bifrost for failure","message":"[WIP] Test bifrost for failure\n\nDepends-On: https://review.opendev.org/750195\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3632512dcafa8adb520bbb4b354f6edc40691663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3632512dcafa8adb520bbb4b354f6edc40691663"}]},"branch":"refs/heads/master"},"980745ef932b2b6fe14baee16e49477789c62a47":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-08 10:40:17.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/6"}}},"commit":{"parents":[{"commit":"11ce4d96d1f206d00a0e58ed718512377771e6f7","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11ce4d96d1f206d00a0e58ed718512377771e6f7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-08 10:40:17.000000000","tz":0},"subject":"[WIP] Test bifrost for failure","message":"[WIP] Test bifrost for failure\n\nDepends-On: https://review.opendev.org/750195\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/980745ef932b2b6fe14baee16e49477789c62a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/980745ef932b2b6fe14baee16e49477789c62a47"}]},"branch":"refs/heads/master"},"3728e7cbb3a4ae8f3a75231e281ffcda827b3731":{"kind":"NO_CHANGE","_number":7,"created":"2020-09-08 10:42:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/750154/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/750154/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/750154/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/750154/7"}}},"commit":{"parents":[{"commit":"6c4f46995a901d1e977c6fba5b540616c781565d","subject":"[CI] Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4f46995a901d1e977c6fba5b540616c781565d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-07 11:51:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-08 10:42:22.000000000","tz":0},"subject":"[WIP] Test bifrost for failure","message":"[WIP] Test bifrost for failure\n\nDepends-On: https://review.opendev.org/750195\nChange-Id: Ic96032b8f0912b36f85824cffdd48f944669604a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3728e7cbb3a4ae8f3a75231e281ffcda827b3731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3728e7cbb3a4ae8f3a75231e281ffcda827b3731"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
