)]}'
{"id":"openstack%2Fkolla~750600","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I8ede183c76a830de06ce6524dc3f6f6944b182c1","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1894240-stable/ussuri","hashtags":[],"change_id":"I8ede183c76a830de06ce6524dc3f6f6944b182c1","subject":"Fix Masakari dashboard policy file location","status":"MERGED","created":"2020-09-09 09:00:10.000000000","updated":"2020-09-14 18:53:38.000000000","submitted":"2020-09-14 18:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750600-1600109526794-c09e01d6","meta_rev_id":"d7ff4ab1feb2719b8c37bd344874d47ea10e4cd7","_number":750600,"virtual_id_number":750600,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 18:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-14 12:56:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-14 17:02:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-14 17:02:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 12:56:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-14 17:02:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-14 18:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36edd26c104bc29201c66d585cf4586a4d67e3c8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-09-09 09:00:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65bd64d80f1af599b04bfee540713aa0bd131a00","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-09-09 09:00:35.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 818870594738baf41c322152b73b8f0f0a623873","accounts_in_message":[],"_revision_number":1},{"id":"579ac91f96eefab1734911a8335a4c247e083bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 10:21:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2fbb72902f4a4651a9562c19e8972463 : SUCCESS in 18m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/78a8f0d50ab04169bda2a40126781a56 : SUCCESS in 40m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e904ec0ac54413c9d9e73f7ddca6181 : SUCCESS in 1h 06m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/52a2f5b4044b49d98f6b7c03e8a29c49 : SUCCESS in 50m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a5e9b444d7e54183a48b0ec9ba945522 : SUCCESS in 1h 04m 25s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f5d044f8330649848da39deebe926712 : FAILURE in 41m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d35bd8124edb4d16972e4b4e38d47e54 : SUCCESS in 17m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dee2c265490a49afa3fb835ded4255c1 : SUCCESS in 16m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e92785cc468e4a61a3be9358e5c97b6a : SUCCESS in 56m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4990c1d7b544acd8d0723e7ac10cfd6 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577ee73f1c814378aeaa11aceed25ab4 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28d78e1486c043919c3c55cc600970bd : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e10110a82e7a4b12b34106bc2dba81f3 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1266f741c2f4e24ab81a64eb53e14f2 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9a0d9597611423f84199be0c7e2c7bd : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/161d9c527c87437b93a898cdffbd37e6 : SUCCESS in 20m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/41f39f326ff24c7aaf2c83f4b789d846 : SUCCESS in 25m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82eeacf42c1844f39c5e8379128b321e : SUCCESS in 57m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b75aa0510874a008f991aa357986d7d : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f922317070ef47b7a98f8fcf841544d4 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"af3d6e5289857e33511c396078b71ebb67add371","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 10:22:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2e4d7e0fc9841599fec7e5698903c16 : SUCCESS in 35m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a2ba4575b9b41ca954ff605126923f0 : SUCCESS in 30m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56f2b34767854aa08397abf11cf2a73f : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1dabe6e0600d4a36ab1281c433dc4ef8 : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7985cf4a123cb6e08e7e97f9db52dd42e350584d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-09 10:26:20.000000000","message":"Patch Set 1: Code-Review-1\n\nbackports must include backported commit id; please wait for merge before proposing","accounts_in_message":[],"_revision_number":1},{"id":"a4209e2ebb7e06675cd84d5bb35e046c54752c18","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:50:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9193469eb626b7b8dcf0490c8e475591e6c421f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:06:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9eb3c03666c841cab45ce06fd4a76f2e : SUCCESS in 21m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3d5f826d023b451eb235908d475d9513 : SUCCESS in 22m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/19053d054c1e4110b1478d64bb71aa77 : SUCCESS in 1h 02m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b66c5cbe21614c73bf10de94497c045f : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/914e4f39be8d4d8c8b51655aec82a552 : SUCCESS in 1h 07m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7725f957a19d4ea59929bb9fda7217de : FAILURE in 45m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a2b6e86ab06e4e5ab2b4f35bcfae6304 : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a8d72b315a66418c8d9d73cc4483c4c0 : SUCCESS in 51m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a5242d553bd4e83a0811346ce4cc84f : SUCCESS in 1h 00m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58da125ece5645cda8ef7bc55ae3a690 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66dcc6e5ed264f00a648e8e9d1c680e4 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0ec253e95564e73b69f72465b7b0558 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fa6f5898fac444b8c61f3a1a9e4c403 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e80e50c62041439ef2f1929590649d : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18803452cd514d5f8f8f41915c6f56b1 : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b3ce28a9dcf4f80a7e0592b296abec3 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/127d9339782f478caee44730920a6ab9 : SUCCESS in 22m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d98c499eb0c140adaa32a0a06daa8266 : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1b84484246e4abd86ccfd93518f1050 : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73fe7e596c7744a0a30eb697a29a262c : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":2},{"id":"dddb688ca53092858db0a701280d2bfaa1cea497","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:22:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5aeba4cac49485c99c0a28d9cbae75f : SUCCESS in 37m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17ea692ee13343d0a5a4d234eeff5158 : SUCCESS in 34m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dd459289e6f40aeb5d00cb6d1b64bb1 : SUCCESS in 1h 31m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cae955921d934cb1a3a9796c374e66a0 : SUCCESS in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15cd92fef81d77650fbdbb61c2f8ce12f2d00b9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 12:56:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"94b5bf52abbe7ffdb01d2762f76aa4094892e43c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 17:02:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c563448a5ed08aa9e776b4000511370ff1293212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 17:03:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"234a5f5d67a2a0a1408e8bbaa27790fa5f6d5fc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:52:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b111b5cebabfacbbee608dfe7f59157f1320087a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:52:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d84ff0adb3e143f79f70c36d83af9c34 : SUCCESS in 26m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/81c7d374a0f544f9add5adeb1483040e : SUCCESS in 1h 03m 04s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/708617375c1c4bbd81c30b64b1ab1e40 : SUCCESS in 1h 07m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a83649ec715a47b5882d1cddc62d1d1e : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38f86f778f94d1ebe4413ed5292b88a : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04190a5ba89c474ba75e0783207a072f : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/578254136178473488148be9c36b365a : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5440151a17e4633a025f12e72a343e1 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baee16af04c44b999f7f67e45778e763 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3d178159503453995b59f1770bc7e34 : SUCCESS in 24m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7b6d54d9ef74f00ae28b9cffcf274c6 : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8a61c0769d514460988bab017181bc61 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e20ef32146c7ea662db74ca8ee5f0032d1b8d089","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:53:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256e3e6b580044e781de55dcefa37b19 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ec8203cd684b389575ac18ead34976 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a31aa02bf428d6ef006e26dc834ebf27c276eda","revisions":{"b5c66044e82e3de9d6a6e5fe112e29c2d6b88edb":{"kind":"REWORK","_number":1,"created":"2020-09-09 09:00:10.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/00/750600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/750600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/750600/1"}}},"commit":{"parents":[{"commit":"8a587c1709aff29ec003813826349eddd3a94b8d","subject":"Merge \"Change sources to stable-ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a587c1709aff29ec003813826349eddd3a94b8d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-04 15:43:47.000000000","tz":60},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-09-09 09:00:10.000000000","tz":0},"subject":"Fix Masakari dashboard policy file location","message":"Fix Masakari dashboard policy file location\n\nThe horizon image copies the masakari dashboard\u0027s policy file to Python\nsite packages, but it should go to /etc/openstack-dashboard. This allows\nthe dashboard to be seen by non-admins, although it fails to load.\n\nWe get an error like the following in horizon.log:\n\nNo policy rules for service \u0027instance-ha\u0027 in\n/etc/openstack-dashboard/masakari_policy.json\n\nThis change fixes the issue.\n\nChange-Id: I8ede183c76a830de06ce6524dc3f6f6944b182c1\nCloses-Bug: #1894240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c66044e82e3de9d6a6e5fe112e29c2d6b88edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c66044e82e3de9d6a6e5fe112e29c2d6b88edb"}]},"branch":"refs/heads/stable/ussuri"},"5a31aa02bf428d6ef006e26dc834ebf27c276eda":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-09-14 10:50:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/750600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/750600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/750600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/750600/2"}}},"commit":{"parents":[{"commit":"ce7cd478449227acd0eca7fdd1670624efc22f3b","subject":"Add collectd-libpod-stats to image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7cd478449227acd0eca7fdd1670624efc22f3b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-04 15:43:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:50:09.000000000","tz":0},"subject":"Fix Masakari dashboard policy file location","message":"Fix Masakari dashboard policy file location\n\nThe horizon image copies the masakari dashboard\u0027s policy file to Python\nsite packages, but it should go to /etc/openstack-dashboard. This allows\nthe dashboard to be seen by non-admins, although it fails to load.\n\nWe get an error like the following in horizon.log:\n\nNo policy rules for service \u0027instance-ha\u0027 in\n/etc/openstack-dashboard/masakari_policy.json\n\nThis change fixes the issue.\n\nChange-Id: I8ede183c76a830de06ce6524dc3f6f6944b182c1\nCloses-Bug: #1894240\n(cherry picked from commit 4f44870376f47c6a65bd02cc83d2ef39fd1f5ae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a31aa02bf428d6ef006e26dc834ebf27c276eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a31aa02bf428d6ef006e26dc834ebf27c276eda"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
