)]}'
{"id":"openstack%2Fkolla~750796","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I12efbaae9c0795f29c2f68158127509fd1b543b0","project":"openstack/kolla","branch":"stable/ussuri","topic":"collectd-libpod-stats-stable/ussuri","hashtags":[],"change_id":"I12efbaae9c0795f29c2f68158127509fd1b543b0","subject":"Add collectd-libpod-stats to image","status":"MERGED","created":"2020-09-09 19:06:44.000000000","updated":"2020-10-02 20:09:27.000000000","submitted":"2020-09-10 10:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750796-1599735068305-0b22d573","meta_rev_id":"091368838fc8d8e75b9826ea950dd0ab8ff90fd8","_number":750796,"virtual_id_number":750796,"owner":{"_account_id":31510,"name":"Paul Leimer","email":"pleimer@redhat.com","username":"pleimer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 10:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":2,"date":"2020-09-10 07:46:06.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-09-10 09:37:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-09-10 09:37:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 07:46:06.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-09-10 09:37:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-10 10:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"154d8fe07f6bee2a3e84e42e8f759067a014af9e","author":{"_account_id":31510,"name":"Paul Leimer","email":"pleimer@redhat.com","username":"pleimer"},"date":"2020-09-09 19:06:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"541759efa8fd34735743fd6db467a0e5202349bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:42:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/152fa096bf1c42c9993be993815e5af4 : SUCCESS in 35m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7853fd37a0ac47259f4218b5d4a70962 : SUCCESS in 32m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f7d48dcefd446ca8daf93739b7667ab : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a3b8ef6de4c47119e810dc36cbdf5bd : SUCCESS in 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71ef13034afd75ed7b6dc42fd02e50af98c6f440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 21:12:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d65faa681c67404aa3aa7c5b4767758e : SUCCESS in 41m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fd3d638bf17b40bdb1a5dcf0ca174c90 : SUCCESS in 46m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1b1017fe2b644478acf6661c40f7c3e7 : SUCCESS in 1h 22m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/960780d7c31640aebcf126732c9682b6 : SUCCESS in 1h 13m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7033b8a7424e42dc81c6ad9dda8120f3 : SUCCESS in 1h 34m 25s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3b397e667f5e4453a911deed6df3a341 : FAILURE in 43m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23c84cc1e48d48989a5330164388683c : SUCCESS in 24m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2d3754bc72b549988951ab0f822ba312 : SUCCESS in 23m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6a73e2dde8049a3b1ebebe359a4cc2c : SUCCESS in 1h 52m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da98d0bf32e54480ac80fee18ccfb83b : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9292bbaf369d47c5a6e899b99ad1ccfc : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617fd926bed54b43b3484ed79011a010 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d16e90a556844e16aa4370e89efd01a3 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39833df66a274495b4e78b6fc698b17d : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/995a759ef6a14c56ad899a212b8fdec7 : SUCCESS in 21m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f81776642414e8998a4870381b664d9 : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5deca1376c4e48619bc96c8c73b6171a : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22fb94e4ee16476a8db132e04bf180e0 : SUCCESS in 1h 00m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2eae3127f83142c38f927962a3994006 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d2f41316ffdddff84787f210e26157d5705c58d0","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-09-10 07:46:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0343afa5047f5d3e18c8ffc1065c862e37f6d6f3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-10 09:37:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39f7652d47d0dd303c7a21a2756a1bd94b530833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 09:38:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a71f6d91f5bcfcda36f80c5fbb61d30380aee8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:51:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e0f4e66270454786a5fc8e96aff7a062 : SUCCESS in 28m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f32120568f854e73a59b975f1c59243d : SUCCESS in 59m 27s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/38e5f85f7c434871a5fd79636e179381 : SUCCESS in 1h 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c80724dc7142c5ac101244ae982433 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29618595f7e14acd94b753ae7585301d : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ba30368f8f342e9a89c424637d3b5bb : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4791e1db0ffe4b8397e2e442ede59101 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226a53aa062f4c00917e6f1df5fc78b2 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59a8af237e5e4850bcf84e38f5ceda77 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9a52a633f91411994a8cc258cbfe46c : SUCCESS in 46m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a142463cbc3f43a1b6e784efcc84695c : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c1b4c08998684e1a0a4eb92675e55cc6d2e6f5ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:51:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9d2af6d75a31df6a3a8f09b462279403571eef1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:53:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472ab993951b4e2ca91e3aaa2e7d9c0e : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e1e07fa44f509a783ae55b1d52c697e61763e54b","author":{"_account_id":31510,"name":"Paul Leimer","email":"pleimer@redhat.com","username":"pleimer"},"date":"2020-10-02 20:09:27.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit be010949b1cfeefdf43151f3c6484c8b20648733","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce7cd478449227acd0eca7fdd1670624efc22f3b","revisions":{"ce7cd478449227acd0eca7fdd1670624efc22f3b":{"kind":"REWORK","_number":1,"created":"2020-09-09 19:06:44.000000000","uploader":{"_account_id":31510,"name":"Paul Leimer","email":"pleimer@redhat.com","username":"pleimer"},"ref":"refs/changes/96/750796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/750796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/750796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/750796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/750796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/750796/1"}}},"commit":{"parents":[{"commit":"8a587c1709aff29ec003813826349eddd3a94b8d","subject":"Merge \"Change sources to stable-ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a587c1709aff29ec003813826349eddd3a94b8d"}]}],"author":{"name":"pleimer","email":"pfbleimer@gmail.com","date":"2020-08-05 19:57:08.000000000","tz":-240},"committer":{"name":"Paul Leimer","email":"pfbleimer@gmail.com","date":"2020-09-09 19:06:44.000000000","tz":0},"subject":"Add collectd-libpod-stats to image","message":"Add collectd-libpod-stats to image\n\nThis patch adds the installation of collectd-libpod-stats as part of the\ncollectd image. The package is available in the centos-opstools repo.\n\nCollectd-libpod-stats monitors resource usage of running containers on a\nhost, specifically those managed by the libpod library.\n\nUpstream: https://github.com/pleimer/collectd-libpod-stats\n\nChange-Id: I12efbaae9c0795f29c2f68158127509fd1b543b0\nSigned-off-by: pleimer \u003cpfbleimer@gmail.com\u003e\n(cherry picked from commit 094a49d41dd1f071cf5f6ac4220611e0ddd69c40)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7cd478449227acd0eca7fdd1670624efc22f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7cd478449227acd0eca7fdd1670624efc22f3b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
