)]}'
{"id":"openstack%2Fkolla~751714","triplet_id":"openstack%2Fkolla~master~Iaa98a8c85d7f105d701d0abfeeaeb8700828da29","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iaa98a8c85d7f105d701d0abfeeaeb8700828da29","subject":"Revert \"Enable mod_wsgi in Ironic API and Inspector container\"","status":"ABANDONED","created":"2020-09-14 10:01:11.000000000","updated":"2020-09-24 16:20:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"46c66c3da2a93dddbd9c13ebdae015c43d23fe90","_number":751714,"virtual_id_number":751714,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-14 12:08:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"date":"2020-09-14 10:01:11.000000000","_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-09-14 10:01:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-09-14 10:01:11.000000000","_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"date":"2020-09-14 11:34:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-14 10:01:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 10:01:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-14 10:01:11.000000000","updated_by":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"reviewer":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"state":"REVIEWER"},{"updated":"2020-09-14 10:01:11.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-09-14 11:34:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-14 12:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22829d8851d609c26d5f908cceb9c83dc1a721d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45eb726f8efce7bf592e0c789e5ce40beed3b8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:01:19.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"fe5b29bc7abce3d1fe8d4b1daedfc182a1fb4c6d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 11:34:08.000000000","message":"Patch Set 1:\n\nWhy? James\u0027s working on a fix. It has not been reverted permanently. Let\u0027s keep it in stash for now.","accounts_in_message":[],"_revision_number":1},{"id":"c52b2fddbeca510a77edae9079903423529688e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:52:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9a5bef84abe4331914075ccb51a0c71 : SUCCESS in 36m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfba2f257a304741ae0588eecace7452 : SUCCESS in 41m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d08fa955caa24aa0b12c214bbd57d32a : SUCCESS in 1h 46m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c187e38b84bd4f59b245ed94e81b630f : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49c96ec991fb3b6e9e4059c49a73b4c005aade1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4b7212a8e4e547e692880cd59b456736 : SUCCESS in 2h 06m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ea3b9af460d949d18f247eb78774e7b3 : SUCCESS in 2h 01m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8426866cb58a47cdbaccfcc6dccd228d : SUCCESS in 1h 12m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3f262da0f42c4ba086ec70eb4f99d0f1 : SUCCESS in 59m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/99aa91500b0d4480b63eb7fc439427b8 : SUCCESS in 1h 10m 58s\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/e1134767b7e24f4fbb0bbf8699792e49 : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ff093b507f1f4e52b5281fad31091ea3 : SUCCESS in 1h 40m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2cbde2bd20e34c6eba7cb41d08d50c23 : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d50eb7ec582c4e1fa9376c5d9792c996 : SUCCESS in 1h 00m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe774e6261dd4359bcc1ea5c1676b2d7 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e657628ce6c442891ccf5c1878bcc5a : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68aa40c190754b9f9155d72d2a31430c : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2b246d050154857ae88c34d695e6789 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f090853d6f9d45c3865f80ae4352adab : SUCCESS in 13m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5688018c19394282bd6eceef7870d701 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/650798e7a4d2434e9b6761da37330d7c : SUCCESS in 36m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93960c0d8d444fcc9fbf79ac9937be2f : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d3105f647e740c2abbfd780dbf6a0aa : FAILURE in 46m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f50d69b04d424e11a8b729ba21767c50 : SUCCESS in 1h 14m 39s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ee0e46fb1af747dbbed95d4a4adf799a : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b61919a08b4ef0033520639725c4bbfd6d2a93a5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 12:58:22.000000000","message":"Patch Set 1:\n\n\u003e Why? James\u0027s working on a fix. It has not been reverted\n \u003e permanently. Let\u0027s keep it in stash for now.\n\nIf we merge the code with a fix then it\u0027s fine, but if not we should not release with these running as root.","accounts_in_message":[],"_revision_number":1},{"id":"46c66c3da2a93dddbd9c13ebdae015c43d23fe90","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 16:20:22.000000000","message":"Abandoned\n\nFixed in k-a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b02d6d4e54513caffe3251ac3baceceb5ebf8c25","revisions":{"b02d6d4e54513caffe3251ac3baceceb5ebf8c25":{"kind":"REWORK","_number":1,"created":"2020-09-14 10:01:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/751714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/751714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/751714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/751714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/751714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/751714/1"}}},"commit":{"parents":[{"commit":"9a02633fafc6ffad03c45c62f27bf8059fd430f4","subject":"Enable mod_wsgi in Ironic API and Inspector container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a02633fafc6ffad03c45c62f27bf8059fd430f4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:01:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:01:11.000000000","tz":0},"subject":"Revert \"Enable mod_wsgi in Ironic API and Inspector container\"","message":"Revert \"Enable mod_wsgi in Ironic API and Inspector container\"\n\nThis reverts commit 9a02633fafc6ffad03c45c62f27bf8059fd430f4.\n\nSupport for Ironic WSGI was reverted in Kolla Ansible due to additional complexity required for Ironic inspector.\n\nChange-Id: Iaa98a8c85d7f105d701d0abfeeaeb8700828da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02d6d4e54513caffe3251ac3baceceb5ebf8c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02d6d4e54513caffe3251ac3baceceb5ebf8c25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
