)]}'
{"id":"openstack%2Fkolla~751726","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I8921befe51da4282121443849177a7ca5ebe8822","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1890567-stable/ussuri","hashtags":[],"change_id":"I8921befe51da4282121443849177a7ca5ebe8822","subject":"Fix kolla_set_configs --check with a directory","status":"MERGED","created":"2020-09-14 10:08:49.000000000","updated":"2020-09-16 13:04:57.000000000","submitted":"2020-09-16 13:01:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751726-1600261308961-b51fe808","meta_rev_id":"3013915ca6ca20ff01be55375e661b68f506b951","_number":751726,"virtual_id_number":751726,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 13:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-15 07:44:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-14 17:02:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-15 07:44:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 17:02:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-16 13:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a74528f5d4c1795e56b8a95e43ca7a9758e49fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:08:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9264f86a1d2a953c9b6d31d727ea80fdf2f28598","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:02:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ea5dab76b2e4266a53c419e613937dd : SUCCESS in 50m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9aa9d3ca799f433998390487b5aa102f : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/677d2e28bf4f4674ab0c61a21813ba3e : SUCCESS in 1h 47m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/72a8c4ac3cc549528c2bd09b1b61174a : SUCCESS in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37062f8b1b3a8c2431e3a742095a3bdf0deaa3a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:58:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/55434c5a220048d3b05d0469fe9c5184 : SUCCESS in 21m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5d521f849ab412c89d31c25c42d4ab4 : SUCCESS in 20m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/294c17b44a484652a367c03423e0b1d2 : SUCCESS in 1h 05m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b3793992cef44190883ca01b0d63c706 : SUCCESS in 1h 04m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8372a24330734b82abb253f20d8f2bbf : SUCCESS in 1h 11m 31s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1f182c39e31244fbbbfa008a30d1577e : FAILURE in 44m 58s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/6aee75372ff9446da80e7ab02343001b : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b49febd44d704aceb795cbbbc9fe5dda : SUCCESS in 1h 35m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/844bde1941cb460e9511b74dd7e609e9 : TIMED_OUT in 2h 37m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6bb41c91782a40ac8827466e06e86914 : TIMED_OUT in 2h 07m 42s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/26a38a55ef1041e1a16a029100fea412 : SUCCESS in 1h 40m 13s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/86c6c37cc97c47978ccf07728a1ec84d : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/43f245a8a0c14d1ab9e8714f0d0cfd63 : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5290eb2fa52647d78716db2ef3d72275 : SUCCESS in 21m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d73561050e540a298599c9098a10763 : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d3ccbc38d1b848e88a979f9e5ba2e211 : SUCCESS in 1h 40m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17252785d5de42be8008bc09c264e286 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c65e3dc56c40e48146246f9071f701 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72001d16d6f04df89099cd52f5437c46 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a02a7845209546058ea2aab9dac8b830 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7a4795218a450a90b979316fa5a3c1 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84fb9539bb8d4d1b9a7814d8dd455631 : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/209dd9bb5cd64f139df65d8a9e12a76d : SUCCESS in 25m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da102f288bee4dd3886916c96d5ac3dc : SUCCESS in 24m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4a99211193e4f3ab1aa187be7a72cad : SUCCESS in 56m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9b4f63861e64eb8854ae690ac9e3347 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2925a7b978c64daf82c24891625ccb6f : SUCCESS in 1h 12m 11s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2782b9a6c98f49a2aaff220711d45c54 : SUCCESS in 1h 38m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d440ae5fdeb140f8a9ece7558759960d : POST_FAILURE in 1h 41m 46s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ed1e3053481f4ec3af7bfccfb25dae8c : TIMED_OUT in 2h 08m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3eb5579c71124087a0e5f604cd1ca7a9 : SUCCESS in 1h 00m 21s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/d8e53446a4554a5e891c76c899d741f4 : SUCCESS in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed80765bab2c90ed82e463f81a08bffe773c8e74","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 17:02:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fa55bfedc074528ede606bcbf0215d05b010d94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 17:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9678029fdf8c5b0aeb84c1d227cabfdb7ef109a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 17:03:15.000000000","message":"Patch Set 1: -Workflow\n\noops, that was premature","accounts_in_message":[],"_revision_number":1},{"id":"222d8361d83662085e2bfe24c35dbb200c5536a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/24e69797de154bfc833a9bdfb25e9bab : SUCCESS in 23m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a4c1d44a69e4a089ac3df74319987d9 : SUCCESS in 1h 07m 20s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/af67423647d842c8a83ba6b9e29c2e47 : SUCCESS in 1h 06m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/858cb0f159c846df8e3343282dd97b63 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f04266cdf784de6a9c3a3930f73e442 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a786c5bd362d42cfbf243a1e7f3713fa : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/162d1ee247004335bea91c0cd4198b4f : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5aac05c95945c9853c79aca52c0a76 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/145858a657c149d599dbb9c3ecf8b905 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d705344097b422f93857be321468322 : SUCCESS in 30m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ee65029bd2c478eacb8f24ea66e1174 : SUCCESS in 1h 01m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/19df5a3fdb8448d49e14f62360297f8d : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"52e6e49d8a27b60b595ed3fbb1b1f79c3934d2b8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-15 07:44:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0bf0fd169306af5b35ad503ddfd0d44e48a84b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:44:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e7002036d5ebcabf0ae20beae905d804cb9d74f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:07:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2c72ce0d8b134f2393102c0ba8f7fa73 : SUCCESS in 38m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b9b6aa99308d419996d188ff557c1f55 : SUCCESS in 1h 11m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6a197be4468f4069bf07e7ca31618ad9 : POST_FAILURE in 1h 20m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9ab874371944d6c8a0236c9298fdaf2 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6994762e02d42c5a1e49b907ae6011f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/915159a226f44f3a8eb535e5b58bdcda : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aeff70e8b0d646f29b605558dec467e3 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cd9b6a18ed4f25804b0eab23baffa5 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/025ed6c5881a47a193527f56eab9c4a5 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/88400c409c604ad486acdc7ac872dfc1 : SUCCESS in 20m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7930f3a5268842e1aa6b67a482e0e84d : SUCCESS in 53m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da640371197f433385af84a5b20323e9 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":1},{"id":"93addda9ca9cf3d18acc783c9682e6767b4f0bf1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-15 09:12:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"330bc07e1d0d7a24f52940d9b984616a75c4dade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 10:36:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/29e11688bd3a420d86e67b62e41416c8 : POST_FAILURE in 33m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79aaf6b4bc5147f89e19c0495294990b : SUCCESS in 30m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd6b3c05ba3d4953893d21060e853503 : SUCCESS in 1h 20m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3c68085d55243e19929fdd0ac12ea8a : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ecc215460220ff825d27bd365732eb0fc94489f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 14:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b6c35cec06f44f06bc0c068ca0b2a603 : SUCCESS in 28m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9f7e138184464c57b4e7a8eccff83b3d : POST_FAILURE in 30m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a22cd375f3ca4f2d9676a8b420d765e4 : SUCCESS in 1h 09m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4693fb25d9e44c05900f26db2d4a6004 : POST_FAILURE in 1h 05m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d001a357d039410f8699491804228dfe : SUCCESS in 1h 34m 37s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/048d3ad319d243c69892d0a818d91c38 : FAILURE in 47m 10s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/fe13515cc37d47a8a2aaa1e0a87a497e : SUCCESS in 34m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/58d90258fa4544eb96fde5713cb28892 : SUCCESS in 1h 36m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e44f14782768486ca71c83587f2f9a07 : SUCCESS in 1h 58m 03s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7a986f7d6c8f486196eff30d0a7e3994 : FAILURE in 1h 35m 51s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9053782f8a4b4a30a0dc062579196873 : SUCCESS in 1h 21m 08s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/aeafa7cc2d5d44438aae9be68a270d29 : SUCCESS in 1h 10m 36s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/c1b3788751a7469596184cd27b474269 : POST_FAILURE in 1h 25m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b56faa27339471ab139fa364dd35511 : SUCCESS in 21m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca1b7446813342f3b72c2285b3bb0f89 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae1b95fc01324ff3997e35f9a118c0d9 : SUCCESS in 58m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/946b965c134845cb99bfcdac0a292aaf : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/040e529e55f54fb39c5243e314197c51 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b72f82b181a4088a3ff08cc12cb53ee : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ae7067f81064c2d99ca3bdf90720a5a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c82984f4bf4aab952fe7e0cc327c7a : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f8d675bf0c94f5095f1c506878cbd66 : POST_FAILURE in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/75d82c4160254f809bd0bc0e274a96dc : SUCCESS in 39m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf613b541136455f8086ae20986704ae : SUCCESS in 32m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b9c4bef8e284a9e8a9f78660dbdec24 : SUCCESS in 1h 05m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/521e55bb26474bf58631d6214fcbf076 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95adfbcd21e14516af7a00d349c78512 : SUCCESS in 1h 22m 23s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/00e1d5828f22419393012cf13dfe8190 : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7078ec9732e14f558d4dbd18c645ea61 : POST_FAILURE in 1h 33m 11s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d8db5cd1f51442c08dd8926908c9e8de : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/bf4c17c3b9b64e6e9a7a38b7c3663ebe : SUCCESS in 1h 02m 58s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8012913827f74f898bc999f1e7f58d52 : SUCCESS in 1h 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9bf6215d96b09abb050a2141c1d75b1d3e2bb06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-15 14:49:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4bf2eefffafe94601c25cbd9300c3c84ebdbb5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f86df0d1878e4fd6a1c8a3613e158878 : SUCCESS in 34m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/228ea8e6469c40a49ddd590fa3b40b39 : SUCCESS in 32m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/849f291b1cde46c28743125300daa141 : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca5a0dad12de4ecda0ddb9c3237ea86d : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"095c0977cdc5237f5152f70b1e51a205eeb0c2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:03:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe4ca94e6f514550ad530aafdfe5fc29 : SUCCESS in 25m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3e7d41215da5441aaec8c7ac8876a946 : SUCCESS in 26m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/986e3aceadb745ffb5f422f90b9bf8ae : SUCCESS in 1h 10m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eef0b308812d4881b914a16f38a87d91 : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/17a73eb5e2f74a9ba10c3a6d06f0df3f : SUCCESS in 1h 35m 42s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2c33f11a548041e68a9a27153b94708d : FAILURE in 1h 02m 29s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/95753e18e35b47ca8c52f0563e4aa286 : SUCCESS in 41m 13s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bd067ef306364ffeaf7e41282c1e7bb4 : SUCCESS in 1h 51m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b9be4b9909894ea29266ce89ca4729a8 : SUCCESS in 2h 11m 09s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f0edf3949a8745d7963b4bb5a101030a : SUCCESS in 1h 44m 08s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/79821ec99abc44688b1ac4a0f9f0bbba : SUCCESS in 1h 43m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/912becde4bf14a0999cabd8967261ee7 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/8bd8f176e0df43488cdc44fe3ca661c0 : SUCCESS in 1h 30m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/712bd5f39c194a62b8751df77ffe2641 : SUCCESS in 1h 19m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/726ed594cba04bd7bafeb40dc19f2bf7 : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/160ab5537f304e29ae76cb421788facc : SUCCESS in 1h 46m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5700c8b5a084dfba30e322465f4ae4b : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777cfffc1ba64936825afdbf1351774e : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5f46970ee704865924afd627898885b : POST_FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdd0ac70d7624b9b99ef5da22be0491f : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27a5f02e4ed4aec83a09233e5d1f1a9 : POST_FAILURE in 15m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0451343c34e4482b8bf803efb139635a : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab2bb090c9424badb2808ff995551ed0 : POST_FAILURE in 19m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/faf3d41f0d5b4981878a3fcfe069a8cd : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/660efbb9f7314eeda37658e50ef5344d : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2a8d40d74d44b618d75c8a10e08f1a7 : SUCCESS in 55m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/222a8c814b264b2da5d7e2a085deb5d7 : SUCCESS in 1h 06m 26s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9247b11a69bb4623b9970c0f57c025e8 : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/afcfce6e4311496594e57e615514bb6f : POST_FAILURE in 1h 33m 33s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/842e3cb4c7044fd5be2cd5eeee6156cd : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a5cd493a75d74ed39d8bda4d9cde97ff : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/af5e85aea98f40ada97835e99c782761 : SUCCESS in 1h 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02f0b58101473f7e5823abfc4a72f98d573b0e26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-16 07:40:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"199bf6d5a4fb2abcd0c77ca70e2cb62bf79a5743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:39:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/296a25fa4fc74e5eb4f3961ef8efce1f : SUCCESS in 45m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7a4c0e1e545448fbf54a48e2c11ddb4 : SUCCESS in 41m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c3590bff68d43ac83fe3ba2f2887e60 : SUCCESS in 1h 35m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/10512bdac0a54c7cb244e76ca6527a83 : SUCCESS in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ae78f47af0e46b55d8cf1d7bcb2cb6b7c96be21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:40:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/427f37776a114ce58a1b752b15245249 : FAILURE in 19m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7297a23edca8457f83e02267f900a123 : SUCCESS in 54m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41be5608a4154067918027568dc18bb9 : SUCCESS in 1h 24m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1d51814e53594d9698f7f8f6872405a2 : SUCCESS in 1h 45m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1fbaa4e751374c3cbc92996dd9dbfaef : SUCCESS in 1h 38m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/30cef693166242f48e307519d0d1ec39 : FAILURE in 1h 12m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/838dd8360d4045c6b15f2e31a0c5834f : SUCCESS in 31m 20s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/34d3c81abf494b2c9de18fee32857087 : SUCCESS in 1h 47m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/598ddd5a33df42559d6d603c72d6e6ff : SUCCESS in 1h 51m 05s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c95d677aeffa4836abe71673e898ee64 : SUCCESS in 1h 47m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/48257bf0adbd4534824de4a69e9a0adf : SUCCESS in 1h 40m 26s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/350ef5b0167b4e4594df1817acc6efd5 : SUCCESS in 1h 16m 29s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/dc19a65acbab4e2182d0db0d9f1f13b2 : SUCCESS in 1h 39m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aacf3a40316346e39a669780eb4c7924 : SUCCESS in 38m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a55d9545933549d1803a7ae63d883712 : SUCCESS in 31m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/688eca4f6344419e91b414c6f9974b65 : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f8bd3fc17604fcc994cce7035e3a9c0 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cf3c7323ef4186aa3b1ca090ef3a0a : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af79bd3bfc0d4a1aaa4929595aedbb37 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad666636cba649fdb36f56d3c40a5065 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2803f259d2b44e6aaf766b5429bc061d : SUCCESS in 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff628df5bac54a13bec0e0ed741c88ea : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2a1c85df80549a2b4925d2b95d631aa : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c5faafa730a4529b1b9a8e02ee7e9e2 : SUCCESS in 25m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/25d0f25464c64e8cac109bc89a6bb8cc : SUCCESS in 1h 37m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/546ba05a90534d1d8e2584bad74a413f : SUCCESS in 1h 00m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27f31c82d1fb4f189fc6b87eecf302b6 : SUCCESS in 1h 01m 51s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c71f1e7f24b8414fbfea4a2feec89fda : SUCCESS in 1h 23m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1a4ca87cf8ff411abae9ab9c2c28e97d : SUCCESS in 2h 15m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f7e0a02dc71744fcbfaf92ae39e1ec36 : SUCCESS in 1h 32m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7c75f3006b584aa8be2eac8bae299395 : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/ec2dc7753aa04bdaac7adb30a621ca11 : SUCCESS in 1h 08m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c0b778d748d21a871621cb01e58d5043317da66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:40:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74b0e2708a8e5ba73efc2a662fc80e22478228e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:01:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/48a19d3c2b4d4a67b185bc3232e79efb : SUCCESS in 22m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f5420933742e45518eb3afbfa2830cb5 : SUCCESS in 1h 14m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0c37b09559ed465484bc9c467cf72b90 : SUCCESS in 1h 17m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54c9306d934f4bc2b679326c26834087 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64ff34ceabb4b20aeab2c788aec6014 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/523fdc933c4e4316b58f133ad73550aa : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdb9d131d5d841948488e0e4cf4d7afd : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bf1d82f70e48fdb1ee5e02a5a47054 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13a227f1c83436b8f040e7da1557156 : SUCCESS in 4m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a90d32ceee14082a40d389759c654dd : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f548af698ef24f92b98e2de22c4ff6b2 : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c08016aca4384f028b4db37e53ff638e : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0892c3f691f491d9e4f9ea92a7751fad7c44225a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:01:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e81673d6137c5e61d6735bcd3cce0bf84015b7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:04:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce21dd64fee54a8b99dc6939a071a7ee : SUCCESS in 2m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9cef3e079cc4951bff1a96780e49bda : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c191e6fbd7ebf0e6f455dc8cfc34ecda64e33bc1","revisions":{"c191e6fbd7ebf0e6f455dc8cfc34ecda64e33bc1":{"kind":"REWORK","_number":1,"created":"2020-09-14 10:08:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/751726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/751726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/751726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/751726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/751726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/751726/1"}}},"commit":{"parents":[{"commit":"ce7cd478449227acd0eca7fdd1670624efc22f3b","subject":"Add collectd-libpod-stats to image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7cd478449227acd0eca7fdd1670624efc22f3b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:08:49.000000000","tz":0},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n(cherry picked from commit c5320eb22386c34d83435fa0b9aa03b178996b52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c191e6fbd7ebf0e6f455dc8cfc34ecda64e33bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c191e6fbd7ebf0e6f455dc8cfc34ecda64e33bc1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
