)]}'
{"id":"openstack%2Fkolla~751727","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I8921befe51da4282121443849177a7ca5ebe8822","project":"openstack/kolla","branch":"stable/train","topic":"bug/1890567-stable/train","hashtags":[],"change_id":"I8921befe51da4282121443849177a7ca5ebe8822","subject":"Fix kolla_set_configs --check with a directory","status":"MERGED","created":"2020-09-14 10:08:58.000000000","updated":"2020-09-17 14:33:55.000000000","submitted":"2020-09-17 14:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751727-1600352963239-b2ad87a5","meta_rev_id":"4baa7fd450a810deea2b9cc1db024af050cfc424","_number":751727,"virtual_id_number":751727,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 14:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-15 07:45:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-14 17:03:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-15 07:45:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 17:03:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-17 14:29:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b802ad7e7c0362eaea80fdda9bc2f850154689b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:08:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0558ca5d5bd67d4a5127272f84b388d31b8bc616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:35:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f3ecc793b4be46c790202e5561b88f43 : SUCCESS in 35m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/46be879777c44d1d9f27d10f9057b26c : SUCCESS in 1h 42m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/248dece3166447949fb0b31ab03251a1 : SUCCESS in 35m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5fa74edbc4c5490fbfb5bcc8fe52e9d6 : SUCCESS in 26m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dcfe6832922e4d1da7218b9f640b4124 : TIMED_OUT in 2h 02m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/01ed4e863edd4257bbec0561ab09556a : SUCCESS in 1h 05m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0d03addf82204c50b7b71cb1ac94503f : TIMED_OUT in 2h 02m 02s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cc6597898425440288fb0ce8c8bdbaf5 : TIMED_OUT in 2h 02m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7339783d85fc42c08d579c095188f306 : SUCCESS in 2h 23m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e7d2cd2d78244e47afe3663f604596e4 : SUCCESS in 1h 01m 12s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a46b22b9c0ea47d0b5002519b5800d43 : SUCCESS in 1h 20m 20s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/32048e4bc2a446ecaaf459ddfa9bb3f9 : SUCCESS in 28m 35s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0937e03a11954273b262c1077e23a14e : SUCCESS in 1h 22m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/766129cf3a3c45d0bd64f4cb90c91d31 : FAILURE in 1h 37m 15s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/c5c5853b523b42518ed63274745c658d : TIMED_OUT in 2h 02m 10s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/aee8cf2fbd214c50bac30c27c916be16 : SUCCESS in 1h 31m 28s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/76713af206c24a79876e474f58b8d251 : SUCCESS in 1h 56m 44s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/860fd7cf8d304434aa2099797fe5fdac : TIMED_OUT in 2h 03m 08s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/119707ac081e49c695d822a377c823ac : TIMED_OUT in 2h 02m 38s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/7aeee08ab3b947f9bbdada91dddbecee : TIMED_OUT in 2h 02m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a878c8130ac94a46a31fc297d70bf93c : SUCCESS in 1h 17m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/24ffdc446ffc4c76a01d155b33f73c0f : SUCCESS in 1h 09m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/84bfbe9fd8b14336967813f2eaaa090d : SUCCESS in 56m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5723b2af95b4c27957c6e7b823d764a : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b88cc85f67e4dd19f46a4719d505453 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5876827c38348b68db67c69c9e1eb45 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bbe92c0486042b09ac40239ec381d6f : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c6324bec85d4a6e91c33d507076900d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a63909772784cbfb9ec44636ecf1c43 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c764b1a87003415e9fc03faf6a110f9e : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d5b58fb432441f58cdc1f694564094b : SUCCESS in 38m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e192a91b66c64b82923254c26ef213c6 : SUCCESS in 1h 26m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bde9b313f96b49b9a1b54bb3a98581e1 : SUCCESS in 59m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71ffe68054cb46839f4a7ec10c53ae34 : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/94f31fadb0d6424196f2d75094c5f5e3 : SUCCESS in 1h 09m 09s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ad06b1f4dd244a4596e04996f27af163 : SUCCESS in 2h 20m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9b3c99eac501459bb785a8d0c0952ae9 : FAILURE in 1h 07m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/918ffea8a903475d808802ca7deeab36 : SUCCESS in 1h 22m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7a093d82fcfb4a598cbb63acf75ad431 : SUCCESS in 57m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5b68c04ec492f71dfd3fefffb7ee0275a570222e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 17:03:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5352966729c9c18bf6a83becd97eab8526c4b1c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/125fb5f440674104a29ccb06617f55c7 : SUCCESS in 37m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/695b6859fdb94a429143c93c6c916966 : SUCCESS in 38m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a147cc87775d46809f9852a06a6b545c : SUCCESS in 58m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9b2b3342ac724cd6a9556a5f1a6dd430 : SUCCESS in 32m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/548816fa9b88483caddb85e3ccbc923b : SUCCESS in 1h 26m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f88f9d8cfe8d45d4a2388c27aa1af56c : SUCCESS in 1h 01m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/05f17cf681f54778992cae34f80956ff : SUCCESS in 1h 20m 47s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/30dafabdbe8041a1aa719d88fe9c200c : SUCCESS in 1h 12m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9c4d70c0359748769ce18e7a350f1383 : SUCCESS in 1h 37m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07b111ab85804d8387d7dd40f200cc51 : SUCCESS in 1h 09m 24s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/00a9cd1425db43f090e1b37239a61e6f : SUCCESS in 48m 44s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/82f28744a8b24dc09ffac57320c7c49d : SUCCESS in 31m 06s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ec574ac314bb4c729369c2adc4200dd5 : SUCCESS in 1h 51m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0189770080e147aa84d66499b44cd7fd : FAILURE in 1h 28m 49s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/1f846489457a4862af5c8d0cd75b13f1 : SUCCESS in 1h 52m 31s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/97a44f9575784ba6869fa012748ba208 : SUCCESS in 1h 33m 24s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8a1055b3587b4c2ea9a72fa25c09bdfc : SUCCESS in 1h 44m 42s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7b548a129e914a4093a5a47f78a49f78 : FAILURE in 1h 03m 30s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/011ff95912a040818fa8fe1391a2081d : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/324a7be6e8f24fdf991ccc51d1874c2b : SUCCESS in 1h 02m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ce589b76cae442c8bafb50184e5f616 : SUCCESS in 20m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/99e3e6797c4e4a98bfe5324fe4db1ae6 : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd82dc7774b74aff84baa5f15e2bb3ed : SUCCESS in 50m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5587db3feeba42b8a281d9f46b0ed27d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db02590d26745d39c4a365cf8441911 : SUCCESS in 8m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8da71fc125c4fc8ba22dbe54dbb6e84 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7be533a7cd4452580c098509d95f23e : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f228858a5f04972a93c5d4ac178fbaf : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a4dc5991fe947a290bca0b29f0046db : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75848d6b3bfb47ac995d24f4c150b237 : SUCCESS in 4m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f441041b2d1f4a9c870de86d946ea2d1 : SUCCESS in 27m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2a8eee368e44e5ea582c3b2b21ebf9e : SUCCESS in 23m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1853a4c12189484aa540bddea3c66391 : SUCCESS in 52m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d26edf8c3c047e49cd81513e7fe1090 : SUCCESS in 55m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70c1310105fc48a3964da7079d528548 : SUCCESS in 1h 06m 11s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/344cfa7e4d7e402db401734815b2e456 : SUCCESS in 1h 11m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e67931ccbb664960941cca42037cf174 : SUCCESS in 1h 38m 24s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a4ab2150d8374082935acf8b70f1ec7d : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ff5d6d5d7b7242f0b6cc158bf02b46a8 : SUCCESS in 49m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"846004f6924b6ee513d323052c8eb809008d7cd5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-15 07:45:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1062fb487c709e3b01d2925823d3d317b9f6395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:45:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e5dca0b657d0293b26578b9ed654ec9b4f89b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 10:18:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/86de5a3c83c94ae191c1d5723b0d464e : POST_FAILURE in 50m 25s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ac9f3ed98734216a106e9894de98039 : SUCCESS in 24m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4fdc9d444d3e49bc87800fe9e658765f : FAILURE in 4m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6221c5dc44b84e88a995e047a70961e6 : SUCCESS in 52m 31s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/55f2927971854287b684d261019e29db : FAILURE in 5m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ad1c148ef5ad4a14a4eefb2b4d699646 : SUCCESS in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72ce29bb3a5b41a1a064b7f295fafd70 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd807db04e241fcbef272690ae02104 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76dd24ea15b945188184338b5e6a1686 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d4f647ab704d1e809c9809cc24db24 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b06bddbd9c0c43a3bcc418347cacc4c1 : POST_FAILURE in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04754872383f4647ae05f8b6847a973a : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/044b5c321be242f998298ae2b952572e : POST_FAILURE in 4m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81b62f8d7d5a4f7abd32a065d3cef5ca : POST_FAILURE in 23m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9210fda0146a40a4ab4d6fcbe469c70c : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02b96c2946d3460cabeb16e788e0cac8 : SUCCESS in 1h 00m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"3cdab624b3274b12475c6063c7ba8df40d88a1e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-16 07:40:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1afe96c6bf84382f688c81eeae1e24d1d28ea015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:56:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aae16175b2284c0b975480ba202658cb : SUCCESS in 44m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/470ed34f41c24b4a977bae624fb91491 : SUCCESS in 50m 06s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/593e9834e245479fae24277b960b9fb7 : FAILURE in 7m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e5bbf30630ee495b942a38c9345e90fd : FAILURE in 7m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7854eaec08104cf0a020a8629ceee0fa : SUCCESS in 1h 31m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/af4c6c3885294c59b01266792acbf995 : FAILURE in 16m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f73c36ff90aa4848b680e3ffb67b3077 : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1f02a26ef66e4660a510260e4742c881 : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e80eb3a6f92245c5b6c001a1296ca259 : FAILURE in 12m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/43df8c0677e94e7fb5b08b6048f3921b : SUCCESS in 59m 09s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/9e2ad7cd8c9d4ba3b691e170e1e4049a : FAILURE in 10m 49s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/e2bb849cd6ec4d68a1106aa2d4edcc92 : FAILURE in 11m 38s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/94e08d3de77e41c7b0cd03aca750953d : SUCCESS in 1h 36m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/fda59fc17ecc40ab8c4aa434526d9e96 : FAILURE in 19m 21s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/89cf04bd986a4d73b491df3d97498b6a : SUCCESS in 1h 39m 06s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/366189fe273640c7a081bb87f1232adf : TIMED_OUT in 2h 02m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/bed0b68d5c394b7a824fcaff4b45331c : SUCCESS in 1h 37m 56s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e5d51b64c9d2447d836c6b6f7a00acbc : FAILURE in 1h 07m 05s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/cd1636ce510741b389e6c61b505447ba : SUCCESS in 1h 24m 46s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/222e07644c9d4f35afd64d331cbffe11 : FAILURE in 11m 35s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ffd808a680e944ad9ab4451d97ecbbdf : SUCCESS in 21m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca48940b847c470f9599480340274a82 : SUCCESS in 20m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0450af008d2047eeb058f293b9d13f83 : SUCCESS in 46m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/922872018baa47d9a06aff30fafac347 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10dfcbf3fb334c4ba6bacac5c5d56b9a : SUCCESS in 12m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f3802eb8cd04e10a2aa2eabae970e0c : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b539d8b62854bae89598a1ee8e01ddd : SUCCESS in 11m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa25a8fc38a04459a48c3fc7a0a6a0c3 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bdbc9d61c84ce2afc7a56a00c7b7ef : SUCCESS in 23m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6711ffc5dee42af8d13f9005ddd85c7 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53c1481c9ab4461bb0d7f653b869a9d1 : SUCCESS in 39m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f62df6d7bef84fe7912d12d66d98fcd0 : SUCCESS in 23m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d9f06df21aa4e5bb15f0c10139446e9 : SUCCESS in 46m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cb6834d8e564e789651dbce4b631de6 : SUCCESS in 56m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca9b5177a4b3434aa110b45e36e39b62 : FAILURE in 1h 04m 36s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/1519de34786c4afeb373ce7040672313 : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3fdd475acfac4d13947e83926d71ad9e : SUCCESS in 1h 38m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/284381f3ae8e4371a9bf0ea85414eb2c : SUCCESS in 1h 19m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3bc1b56e5a514e318ea3a9ed9aae95cf : SUCCESS in 1h 23m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"90a492d4530dc785f53a69bddfe7ba3ca48131cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-16 11:36:41.000000000","message":"Patch Set 1:\n\nINFO:kolla.common.utils.base:  - Status code: 404 for https://repos.influxdata.com/rhel/8/x86_64/stable/repodata/080344777bb052c0569e6c66a064b39298999826-primary.xml.gz (IP: 99.84.251.41)","accounts_in_message":[],"_revision_number":1},{"id":"24331fcc7636046fc95d6361b08c88dfbacd75b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-16 11:37:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46045e8021604c06ff2c5f34e04c5058ccb47089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:30:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a5b9ada75ef44841977b71b0a57802db : FAILURE in 27m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c5293931f550445194b87636d85f33d6 : FAILURE in 21m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bea8510d980d491f92554dc840b65797 : SUCCESS in 25m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f29f0ae30db64cfc8f13bcb26ea43bf8 : SUCCESS in 27m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4a15331313ea4f07af6e1f78646062fe : SUCCESS in 1h 23m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/771c7066f4154338948e33e627f98a0b : FAILURE in 14m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7d8508563f3645d1bc524f95f095a58a : FAILURE in 12m 21s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/853a1c42e9004d25bc194c13a738dcd8 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/520bb9702a15445aa5e61d10bc7de982 : FAILURE in 13m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5614f163dc0b430ea94f03d2b73403b4 : SUCCESS in 1h 12m 59s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/31d271f14d0b4e268a9a0ad5bf2d7cfb : FAILURE in 9m 25s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d013c70d1d394d1aae8cbbb97aeb4a3b : FAILURE in 13m 38s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b8e7080ac38f4a9982db6a1520d514e1 : FAILURE in 12m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a53c930416504b68b661aa193c7cf8b0 : FAILURE in 1h 10m 57s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/afdd136a34e146409749aff2c0a12dcc : SUCCESS in 1h 45m 33s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a3935854e2a54cdfa70661288ca73244 : FAILURE in 12m 57s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/55506e5c1348428a9bc799c8ff24f0b2 : FAILURE in 12m 33s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e1106c35867145fd938de09dc795c339 : FAILURE in 1h 08m 22s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/15cefeb3b3d84d27845b429b2f2ea867 : FAILURE in 13m 06s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8a9b1a69d5734ba9a5e4364a67503673 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/54c4605a23fc4f369b6e3fdca1b5080f : SUCCESS in 23m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2ddbdd4c61a34243a84ae24f4de11db6 : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82e0f2a4d5d444a89f3717b2b06e2c02 : SUCCESS in 56m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be1fee4a846244b7aea9cea460916956 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd7cb28ea7640b4ace22c0068968099 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2e166cee2b842449ecd708dfec9de28 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e5a0d3f5d2e4d118a58c8b68d5e5306 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/571bb26d33ec436497ccf729d336ba2c : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad424626c8e412390b8384069afe925 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f9e2fbc16794e629f90745093d6cf9b : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30e3f409db2049a6856887677a91e557 : SUCCESS in 23m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ad43dd572054f7db25e743af975670b : SUCCESS in 24m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19b8f6e371404e7a80e4dcf7cc54bfd4 : SUCCESS in 44m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1a135bb6a0149b1bae3df6e8953d3c9 : SUCCESS in 54m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82be57ff93da47f4bf44b35b0557f028 : FAILURE in 1h 01m 26s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3f1cf346d82e495e85ed8e8244459288 : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0871a9a7735a4b499889973a5be85bcf : FAILURE in 1h 33m 50s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6cda5ae30d5541d096abb368f87fba1d : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cc10ad65285045999b44cfeb148351d2 : SUCCESS in 48m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"2324cae5d85951ccee9dd124f6c05f893f86b97b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-17 10:02:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"101585267fee563a86892a1c01d87f239d8b05bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:41:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/378232c0d23c4c26bf37a2fcbc12fd64 : SUCCESS in 45m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/54b99bbd61b248afb035bec7eeeb75f6 : SUCCESS in 48m 57s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fbfb8e0aa9a44dbf9a8eb37c68f0ec7c : SUCCESS in 23m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c3a5bd49efa0400ba754d3b8d00b35f3 : SUCCESS in 34m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c7307a2ab4ac48548046ac32a7700189 : SUCCESS in 1h 22m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/06c97ce872c14c24be33e055de3c215d : SUCCESS in 52m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b1a2526ae7a34b55af9860e798169b6a : SUCCESS in 1h 41m 06s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/486ba92cfc634d4ab300e6729871a598 : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a9ad6c0b1a534f678e9c953ec11d9b7c : SUCCESS in 1h 20m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9afc3ffc988b4767bac4d0bdddff9ad3 : SUCCESS in 49m 27s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2ff9210546cd4e5fb03e1df666ae2475 : SUCCESS in 41m 55s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/8c9f0e09d627476195afc69c7283c0ed : SUCCESS in 26m 37s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c194ade44a484003b4286d0ad86b26ad : SUCCESS in 1h 58m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0223a78b64234c4489c795ad9620d87a : FAILURE in 1h 19m 27s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/9b090b85932443fc9e4d16ff7e6e3889 : SUCCESS in 1h 50m 37s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/380f73d34e2b4ed591dcb77dc97c2e4a : SUCCESS in 1h 28m 32s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/05f4b16a80f54115938f2e462778bbe3 : TIMED_OUT in 2h 03m 25s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/878231cfe72d44fb883a9add84edd1af : FAILURE in 1h 20m 30s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5a58387c2dc74bb593d981d63dcdf1db : SUCCESS in 1h 24m 25s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/07a5a3b67a014fe08dc5eb8b7d3300cd : SUCCESS in 1h 08m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a4605b078f034428b4d12fa0a6302275 : FAILURE in 6m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b4485fe8d7514b65ab2fd1744704c006 : FAILURE in 6m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8af791d29a94a7392caed6e41caec41 : FAILURE in 10m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f4f33bf6262492380c9086846820f8d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36decc0b4d0e4bd487ac0c63f8d109cc : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14088baa02d843969a6c54819c525f80 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11934574cf184f08b38430dc14517d0f : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04428616b46b4f04bbcc6eb290dc78f5 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe496e3e36684adf9c35b95da5be2fa3 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dff270d88ecb48619cb9827a2687ddc7 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32bab4ad99e6456fba41aa9ee4b0ffe9 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a033c2ae5b74f4083329b2f2d962bbc : SUCCESS in 22m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a908bb8a61a4767adb8378da2eb67e6 : SUCCESS in 1h 00m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2a569eeae214934b929ca56fc588fdb : POST_FAILURE in 36m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b43d363d9a7e472e8848fb0ba1fd9882 : SUCCESS in 1h 15m 30s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e0a71be6723842ccbd542ae3f94fbe2d : SUCCESS in 1h 27m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/df477e7da7174c649d4f0b9d92b229c2 : SUCCESS in 1h 30m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/aec9354216054d869b397814cbd784ca : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f5298b64da60478b9193221ab6db2982 : SUCCESS in 45m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d1da45da6da814c102934c9a1d7b90d0132b3305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:41:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef2e863c65b29ef80113051df261277c7924026e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:29:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e68a01f3045248ff89f71b74d756aac3 : SUCCESS in 47m 54s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a9fa237838284ff591e5c64195f2c8cc : SUCCESS in 29m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dcd9c53d40c9469a8db0907713e2550d : SUCCESS in 1h 45m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/403b3f4770d44b519df354ea2a27ffac : SUCCESS in 1h 10m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4052b05eecfe4926a1f2f96cd5b5786a : SUCCESS in 1h 29m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/19b2c1b246de485f890e87467d579148 : SUCCESS in 1h 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4452e9443012439eb6c301635659f594 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b5b5824a8e49b396b929713cb74d05 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79377480612d44cd83d935cd3d44b4e9 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb09b3d0a4994509968742a85e7113e2 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66186d7410e04ed6beed34be6768dbf7 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7979c9300bf44c7cb746703e87bf19f2 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b25d892f81c946fc8a6a7d9f7e72b7e8 : SUCCESS in 4m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed34b4aedbb4466683158c839f7eabb2 : SUCCESS in 29m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcae9651f7ff4546b38812fd2ef1d35b : SUCCESS in 56m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cca91ff4bcd441b1bddf563b83fde9b9 : SUCCESS in 1h 03m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"36c2913b27dff279e18c2e25dccb52d3bf649113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:29:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da3a311a0df72ab51eda601ab735549814e2e8d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:33:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32480a0703345898517bc6076ea1731 : SUCCESS in 2m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61ed3a1e691c445dba985635d6e319f1 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"797ecd6ddbadcc7a726ce1611c210667565d8a91","revisions":{"797ecd6ddbadcc7a726ce1611c210667565d8a91":{"kind":"REWORK","_number":1,"created":"2020-09-14 10:08:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/751727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/751727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/751727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/751727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/751727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/751727/1"}}},"commit":{"parents":[{"commit":"5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6","subject":"Replace cradox with rados","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5826e44d7b0f00eeda9dc3ccb68f2558a61a49f6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:08:58.000000000","tz":0},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n(cherry picked from commit c5320eb22386c34d83435fa0b9aa03b178996b52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/797ecd6ddbadcc7a726ce1611c210667565d8a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/797ecd6ddbadcc7a726ce1611c210667565d8a91"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
