)]}'
{"id":"openstack%2Fkolla~751728","triplet_id":"openstack%2Fkolla~stable%2Fstein~I8921befe51da4282121443849177a7ca5ebe8822","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1890567-stable/stein","hashtags":[],"change_id":"I8921befe51da4282121443849177a7ca5ebe8822","subject":"Fix kolla_set_configs --check with a directory","status":"MERGED","created":"2020-09-14 10:09:05.000000000","updated":"2020-09-15 08:39:34.000000000","submitted":"2020-09-15 08:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751728-1600159089989-a980bdbc","meta_rev_id":"83bb4cab1405548d8060d25370739380c5438508","_number":751728,"virtual_id_number":751728,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-15 08:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-15 07:43:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-14 17:03:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-15 07:43:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 17:03:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-15 08:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"219b31969ee532f7c8f6426117014203f9b64681","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 10:09:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da3f9b9f98a3bd0bac9b413e818a8636e3c24c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:13:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/064624a2651447138b5401b4d859c1f9 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/175c977df8f44705b528c73992a9e0ea : SUCCESS in 1h 03m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0b26c3422deb4c7c81fa652f407a8942 : TIMED_OUT in 2h 01m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7ca5e88f97c94027b0945490e5614ade : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c3a4622d4da246a298b9b23d3799aef4 : SUCCESS in 1h 13m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/02940701aa9e442d8494048eb40bcdd8 : SUCCESS in 56m 23s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2ff09d6893b94d1297a04cd79fce7b90 : SUCCESS in 40m 41s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/613ded76995f4445a74b415f9c23ff37 : SUCCESS in 57m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78fb973316d345fa963b402a3de29205 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4cb78d10124463eb139bf2332157e55 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d3715643be44a54bd5cd5f9853b7978 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feda5059cf8742b4ae4dfd10ea065387 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e94ee1bb51de451baa3475155c97a6d8 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643dec98d6b34a2f9a8634520e881ca3 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af748e57c6ef4e50bd42cc7c90d4e1f2 : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/723434ba201a4988a4f57b1d71aae4b4 : SUCCESS in 38m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb9dca7d84e14281b06e79901fb92a23 : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/183658d0ebb347a9b563b02fa38c4400 : SUCCESS in 51m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41e28fc137ad4b4eae9695c00044455d : FAILURE in 30m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/206cf9c218984e909abfcc189c99c509 : SUCCESS in 1h 09m 37s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7f442df8fa9e3f40808f5c0ef6d75132c33bb1c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-14 17:03:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6afeeaf1a06a238bb3b4c15f46d3d4777f7a3a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:30:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2662bfde8eeb4885a45e1efba596ef8a : SUCCESS in 37m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f20baeb1a47547b4806d1d549e6e5ccb : SUCCESS in 38m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b7dd470d10ab4e31b3397dbcea6b3521 : SUCCESS in 1h 19m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d093cd766f794186831c0ef015e58113 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36009cba2e4140d6beb8f9fc894250de : SUCCESS in 1h 07m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eddcf5e9186f4f439694de836316ef4a : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9887035c5f4148d98673434da36e1bc1 : SUCCESS in 49m 33s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b6d19edf618c4679a68b0ef38c8c249b : SUCCESS in 42m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5565119df8a1497a8ea5d0611c82a7b7 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96214bbdbc0455fa791c85275eab79f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91cad310346f43e88be2a4070aa2a8ac : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9014908ded554cb8a05246e7702f1a37 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a19db1bc26547808413e2c4fcfb27be : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575447d758a147aba01dca1743c18b4f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55e8cfb59f304f229f9b804e112363af : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88f2aefff4b743f0ac375e973e8bd375 : SUCCESS in 27m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d59761800a424e4c899f56cd181408f9 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec3aab553618441d8827b499dcfdfc67 : SUCCESS in 52m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0fbdc64bf8b4156ac3ef529302a9345 : FAILURE in 33m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d241b2ef2a94f51a594d02038875818 : SUCCESS in 55m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5dea5c442dd37bc6b8652aacc600daa6b3776f5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-15 07:43:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e488c7ce9943264db7e6775d90e5849793a8cbdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:44:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29c04aa5cd0905a8d7ee330d6355b00479649983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:38:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5928a6b0d5324b090f547029bfcdedd20580472b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:38:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f0c7840044d745f2af3ebb8eb2269704 : SUCCESS in 42m 35s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7278bfcba96a4dbeb86e5554ced80bc9 : SUCCESS in 47m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d39d98c5d8d4ea1bf2d3511fcd4c0f1 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560871271fb3412a919ecccd8ef08e12 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa269cb2a6a148c4b18077817f9a6c9f : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb75cffb0b724e4586ad93d57f7c7fb5 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a78412b737d42a2a7c2ee2916730a78 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd52c1ac9214482daf2cc26ff147d80d : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e8a8f43e2cc4a1c80b42abde72c85ed : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/57a87f550fd740ccb4ac2c9a1c456389 : SUCCESS in 27m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5b4bb4d5a4d97d7a0432c4a35599eaebe34daef7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:39:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684a994dd91744eaa2b13db20685b512 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/992f5316811d46ba95651785459efffe : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6835b12a6f1e8f715aa5c69383a8d1836551496","revisions":{"c6835b12a6f1e8f715aa5c69383a8d1836551496":{"kind":"REWORK","_number":1,"created":"2020-09-14 10:09:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/751728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/751728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/751728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/751728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/751728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/751728/1"}}},"commit":{"parents":[{"commit":"a34ac859df196a7eef3a7ee6bf4d954841dcf313","subject":"Add glusterfs-fuse package to manila-share container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a34ac859df196a7eef3a7ee6bf4d954841dcf313"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-06 09:50:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 10:09:05.000000000","tz":0},"subject":"Fix kolla_set_configs --check with a directory","message":"Fix kolla_set_configs --check with a directory\n\nThere are several issues with kolla_set_configs --check:\n\n1. We calculate the destination path incorrectly when comparing a file\n   in a directory, due to passing arguments to os.path.relpath in the\n   wrong order\n2. For directories that have not changed, we also attempt to compare\n   them as files, which fails when they are open()ed.\n3. If the config JSON does not have a config_files key, it fails with a\n   KeyError.\n\nThe first two issues affect the fluentd container, which specifies\ndirectories as the source, without using a glob. The third affects OVN\ncontainers.\n\nThis patch fixes these issues.\n\nCloses-Bug: #1890567\n\nChange-Id: I8921befe51da4282121443849177a7ca5ebe8822\n(cherry picked from commit c5320eb22386c34d83435fa0b9aa03b178996b52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6835b12a6f1e8f715aa5c69383a8d1836551496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6835b12a6f1e8f715aa5c69383a8d1836551496"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
