)]}'
{"id":"openstack%2Fkolla~751787","triplet_id":"openstack%2Fkolla~master~Id0e738044a1931f9d611a7281a48ea4a593f1cf1","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Id0e738044a1931f9d611a7281a48ea4a593f1cf1","subject":"Improve pip install process for offline deployment.","status":"MERGED","created":"2020-09-14 13:27:05.000000000","updated":"2020-09-21 09:34:43.000000000","submitted":"2020-09-21 09:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751787-1600680695620-2eec13d7","meta_rev_id":"a5219494c34ead1660ca6d36b170823abd7230cb","_number":751787,"virtual_id_number":751787,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 09:31:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-21 08:15:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-18 08:50:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-18 09:21:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-09-21 08:15:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-18 08:50:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2020-09-18 08:50:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-18 08:50:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 08:50:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-18 08:50:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-09-18 09:21:08.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-21 08:15:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-21 09:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f07b2a6cdbcfea06790c3d30c9dd786d6b41ede8","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-14 13:27:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53697ed53a510f67e3fa28204573f1af77e029ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f260f561c56e433ebcc4e8a5f2d0bcf2 : SUCCESS in 37m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b832df1ee8024423898fa2495ea42100 : SUCCESS in 36m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1db562dc70184ff189f0156563408b70 : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/606637213c314aa78227bc76804d9ac3 : SUCCESS in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc48d8532043c5ff1f039d663c3d1be29c92a796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 17:47:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a3ba9189607b4d258bc0af2fd4bc0083 : SUCCESS in 21m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a92ebd3c32534e8481e1cf4a4b70942e : SUCCESS in 22m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f37ffe1c56dc4bb792fcb76c5ca54a57 : SUCCESS in 1h 04m 36s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/37f8fdc9757b423eb865460525ee35bf : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e0f4f357a9834fc6b01a101b8678a938 : SUCCESS in 1h 21m 36s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/26398c523a514534bda1b208491d843e : SUCCESS in 36m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ab894a79500540b9b652f602ebf432fc : SUCCESS in 1h 33m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/62dd1f5e08054ed5bc9a2a2cee3c5282 : SUCCESS in 1h 47m 00s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6f72d9c7458a43618f624739782004e0 : SUCCESS in 1h 58m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ad95827dab1f43c1895e1ac627a58a6a : TIMED_OUT in 2h 05m 38s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f320b12d4bf84d0a8e98fc72a1fc7f6c : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/abd8e2d339794c22b5c1fbee0b485018 : SUCCESS in 1h 29m 16s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/4569ec78bb9c40f6ad6025a0c417df57 : SUCCESS in 34m 55s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/990c43981690492aade6a8586c08b0aa : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/a3a354dd4681473088751b70fce509f6 : SUCCESS in 1h 23m 30s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/7d6c852547a2424b9f358c1cd80e2a46 : SUCCESS in 30m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5aabd0e4bcf4d95be1383a3f4a44b90 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3d42a756a1a4b709f7de2766d06c33a : SUCCESS in 17m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/44b9a7714f5b41d8b7b3c8755c92dce0 : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e50b18e619040379cbcf53cce4948cf : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e47a8f2adba41f9865f5fd5d99b05e2 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6f41ed6054480098c1ecc6bdfe3d8a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c7a3b13801d4d02885014c4a2ee56bb : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde7a9c847874570910f589680b41e36 : SUCCESS in 12m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3052239f0c974f1b97fd766e4188d53d : SUCCESS in 19m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/329f43a9eae34d4c9e317633475f4505 : SUCCESS in 1h 11m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93754bf48a7942d1be12d5643715c475 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9fac64b1e31e44fe8b85ce7332bdda7a : FAILURE in 36m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f11b31eda3244951b87511f4542321c3 : SUCCESS in 1h 02m 33s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8ec07a6615474feca9891974baa5c0eb : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1b91c61b1d114e0e961e339ec978608d : POST_FAILURE in 1h 36m 38s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/dd50aac8b962455098e62dbbcc3f7405 : SUCCESS in 1h 16m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/014122bdc7964c0890e9f0d3933f9709 : SUCCESS in 47m 57s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/0510fad58a4c4c44a43f80cabfe8c674 : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/19c70e6221014bed989d6deeb28139b5 : SUCCESS in 30m 11s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/b0b76d6ec8a44b64866c8fac9ab65a8d : SUCCESS in 47m 59s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/f4817c31f46148f8827788205c535dc4 : SUCCESS in 48m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/ad2feee26ab74340be80a91fd51c8092 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/49ce56efdf45467e9301806c872e1181 : SUCCESS in 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"641332bf2b1e78c5acad62986725cba7b7b8a1b6","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-15 13:26:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b546d732a427030e9bd45e7a3b99113707114f9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 08:52:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"64b82a7f11c58ae8525d5fe436eb5117f05ee79d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 08:53:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6c24769e46627ddfe50b47ffafec3efe9b6e61f","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-17 10:25:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0107b8d273261ae7f4c83fa174ab151d75ebd237","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-17 11:24:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\n\u003e (3 comments)\n\nkolla does not support centos7 in \u0027V\u0027 cycle  ,so we don\u0027t need to think about python2.X","accounts_in_message":[],"_revision_number":1},{"id":"5eca94c17657f75026fbb2f6e3a247c8d55b20e9","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-17 12:12:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6cff1e7b7ec24d3e5f5c6f011560fa8600f797","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-17 12:41:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fad5720be73e6182b71b6205d5f2ac96c5c64d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 13:33:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/60135c6819b1451baa8bba785f248e76 : RETRY_LIMIT in 2m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/833652ee18f043fb9075152acc04ed01 : RETRY_LIMIT in 2m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/00150f8dd4a74015a93c41910736d478 : RETRY_LIMIT in 2m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e14d04c59c644db6b4bdfe375c199566 : RETRY_LIMIT in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65a04e6c2655015a8869795116f875cb3f98df78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 13:50:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4a05a18e05cca214affcbdbbbbd323e00a0bf578","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:41:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/642b53aab0694cefbd72303b83d96ed6 : RETRY_LIMIT in 2m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03100e27b6e543e8aad184df8f6450eb : RETRY_LIMIT in 6m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed29e328e98148edbd746894675eb577 : RETRY_LIMIT in 4m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/100600bd86144bd7a74ce30bc3bbaec9 : RETRY_LIMIT in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f5e0d3c503dec8b6974484a4a34a840f132cf2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 16:15:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/48492e9c17244bd784eb637887c3403d : SUCCESS in 25m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ab608cb4f3ad44fda33bb3f9c49b3f50 : SUCCESS in 26m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a67fbb1538942bfbb906131dee210a4 : SUCCESS in 1h 33m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c3ef3483a64a4e8baf1b44ba9bcc2ff0 : SUCCESS in 1h 28m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1b66d2ccb70249e8b0b60b0ef563be65 : SUCCESS in 1h 11m 45s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/64e07afc6056439cb29e10083fa517c8 : SUCCESS in 29m 38s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e9f03923fbf94cd49c3624b60bec4f28 : SUCCESS in 1h 16m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1f13655288ae437aa65b436e99e1a29c : SUCCESS in 1h 36m 05s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8d6aeafafb504e97bb2d32e1a68b05f8 : SUCCESS in 1h 25m 19s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/82e9e01099a641b19b983b25d34a502d : TIMED_OUT in 2h 05m 46s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/d4c9648675b241a7bb007f6c2e04f0fb : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/33a430576c5943ad9b1af356f1de913c : SUCCESS in 1h 55m 55s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/ac7875c519f74aedb5fc4d4d23be1335 : SUCCESS in 38m 24s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/6442f93fc7c243e680dc8e3f88688bd4 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/85143808703e42c1be70b49eb86812ca : FAILURE in 46m 32s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/ca2bc56b6e83461fb9b15d3b8a8040ee : SUCCESS in 42m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/599aeb4043ae45ea958bda900135d2c6 : SUCCESS in 33m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d915c0e975e64f29bcf6a85e7977adf1 : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ee83d72eff54c6eb31dfc1f82aed727 : SUCCESS in 51m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32b5ed1804c14210a84d8eb282a71ba4 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3a67513c64405fbd0bace754a9dc54 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4986a3bb02584e05b3ae4c88e0ce67af : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc9ef35507484c31ab22d6a50f5cb0d9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba2eba3fa9a445e9044e502862a936f : SUCCESS in 14m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/314af8443ed44388aca2125799d6388b : SUCCESS in 22m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b737090586e443c1afac1aac4342cce9 : SUCCESS in 42m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75dfbd0427934173ad246915ee3eb42b : SUCCESS in 1h 00m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9fe3c6daeb44325946dce1a3e406b8c : FAILURE in 47m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3754142f3c7647eb8bacf52d4d8c510d : SUCCESS in 1h 54m 59s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/cd78f19ddc264c74a5740e56e88c1322 : SUCCESS in 1h 17m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/eb9104868ffd43d4a3b57caca0d50923 : POST_FAILURE in 1h 36m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/72ababed55ef489787b4af192e9158e0 : SUCCESS in 1h 34m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/acaaeca30fac4a0da9ebc1b84d8e5331 : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/1494b7f3f66d466c9d79a17db2dd6020 : SUCCESS in 1h 27m 36s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/99a50cb31b6a43e9a4320402f23ad0f0 : SUCCESS in 39m 51s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/208a76f4f97d4b6f82579d9f83d766df : SUCCESS in 45m 38s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/3a9de5e4a9334d25923a25c1ecfc54ce : SUCCESS in 46m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/a358850d28094449ab29fae7c6a04596 : SUCCESS in 1h 38m 06s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/15aafab6c5ba443ea85f71dd00e69e0f : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65f10473eadc6775a2efaaff9259b97c0c7d6981","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-18 02:40:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e64c7d02fc0664427ff7dd45f35d356520b83984","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-18 02:48:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7c18e978404b3b3f8e5156340a33d0b9049631dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-18 08:50:17.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a0f88f7cf70e3a36efd8008c99f534e0be34292b","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-09-18 08:51:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a07a6eb788c4505bb9c55e07101fe328363ce4b8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-18 09:21:08.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)\n\nif we compatible with train ,then we will compatible with python2. upstream seems we need drop the support with python2. \nanyhow this ps looks work properly.","accounts_in_message":[],"_revision_number":3},{"id":"ccb6c0de6427305cf9c33be9ed8034dc4a0ea0d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-18 10:18:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32551186932ff5fb264ec41302297c4c3db10c2b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 08:15:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"4eb1b0d9a26e86edd505590f0f1044105d0e4364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:15:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f0d49c8ef548acb5e678373c6656f85a557208b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:31:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c6ec83b4a192637f10b9989d800ab68aef2623ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:31:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2d69590390c04d07bf3586929c4406be : SUCCESS in 23m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0dacd4aedaa349c9b0daa2c1a82f7d1e : SUCCESS in 1h 04m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d7aa1d57c3404738b57c0c16876960d8 : SUCCESS in 1h 08m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd7cbf92c4b442e2b05d7eff3e82ed39 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41aab6be964490a8e94b9928eb31267 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79a443c4458c47c3aacca337821834bc : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7236628b3840289ddf0e9048a2152d : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d037ffa957a24f97bab786885463fb16 : SUCCESS in 16m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cd7ac3406da42a4a26527d30dc94f49 : SUCCESS in 22m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18a90d5975af4bda8fa4a3c759123689 : SUCCESS in 46m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2496807bbdd5414e8cb1a6a8f0b8f6fd : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":3},{"id":"a5219494c34ead1660ca6d36b170823abd7230cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:34:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ca9625e6f44159b152bd7b349d037e : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b93c40a3635fbce72e48dc29120e50484cba36ca","revisions":{"1b18476ce01ce5822ac47ed01844a59df8804bc5":{"kind":"REWORK","_number":1,"created":"2020-09-14 13:27:05.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/87/751787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/751787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/751787/1"}}},"commit":{"parents":[{"commit":"0cbba11213f3dcbd9a58815a3e9cbab5c0597faa","subject":"Reuse the docs deps for releasenotes to benefit from constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbba11213f3dcbd9a58815a3e9cbab5c0597faa"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-14 13:26:41.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-14 13:26:41.000000000","tz":120},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n* \"Use distribution-provided pip\"\n* \"Set pip version according to python distro version\"\n* \"Install pip earlier in order to avoid multiple installation\"\n\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\nCloses-bug: #1893204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b18476ce01ce5822ac47ed01844a59df8804bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b18476ce01ce5822ac47ed01844a59df8804bc5"}]},"branch":"refs/heads/master"},"a645393aea1f95159c53c34fb8e9479df31ba728":{"kind":"REWORK","_number":2,"created":"2020-09-17 12:41:47.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/87/751787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/751787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/751787/2"}}},"commit":{"parents":[{"commit":"9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b","subject":"Merge \"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:41:12.000000000","tz":120},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\nCloses-bug: #1893204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a645393aea1f95159c53c34fb8e9479df31ba728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a645393aea1f95159c53c34fb8e9479df31ba728"}]},"branch":"refs/heads/master"},"b93c40a3635fbce72e48dc29120e50484cba36ca":{"kind":"REWORK","_number":3,"created":"2020-09-17 13:50:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/87/751787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/751787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/751787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/751787/3"}}},"commit":{"parents":[{"commit":"9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b","subject":"Merge \"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 13:50:15.000000000","tz":0},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\nCloses-bug: #1893204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b93c40a3635fbce72e48dc29120e50484cba36ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b93c40a3635fbce72e48dc29120e50484cba36ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
