)]}'
{"id":"openstack%2Fkolla~752434","triplet_id":"openstack%2Fkolla~master~I9e4e408ebc197b12ef128ec629fb648246ef6582","project":"openstack/kolla","branch":"master","topic":"kolla_toolbox_ovs","hashtags":[],"change_id":"I9e4e408ebc197b12ef128ec629fb648246ef6582","subject":"Add openvswitch packages to kolla-toolbox","status":"MERGED","created":"2020-09-17 11:13:42.000000000","updated":"2020-09-17 18:59:32.000000000","submitted":"2020-09-17 18:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752434-1600369061657-309c905f","meta_rev_id":"fe0e4bc62641c839ccfe95eaa32a546a0f3fcd81","_number":752434,"virtual_id_number":752434,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 18:57:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-17 13:16:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-17 14:16:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-17 14:16:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 13:16:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-17 14:16:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-17 18:57:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97d52b650a68292a20ca8cd31f0a2e7c7eeffe42","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-17 11:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc60f32e43e2c5c4f76ac97dd2a504ab5dbcb40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:28:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7a1c7e0f9a834b5f9fc269243d511d8e : SUCCESS in 25m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/40a3a2d8a8f3462383f1e0b6c3b8d350 : SUCCESS in 51m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/58a1252d95c84665b5ffe7c457d9c503 : SUCCESS in 1h 10m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8da5f4fbfb3d42da9723e92287bb186c : FAILURE in 48m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c22ea091a8d748e79b2fcc62e8cf4a2d : SUCCESS in 1h 08m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89d4f3761ce74194bf68482dd412745b : SUCCESS in 18m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/53fa8e50d42a4e2f9c39e1f4c389ef4c : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6df0ac9d30e544208fa2c8d7030b52ee : SUCCESS in 47m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6e59df190f54dd3a977769cee1e708f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c362c1df8946d0af3b4711470b00f3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d096ce9298904578ad5eed187090e90b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94213f57749f48c8858a5768ba0d83a6 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c6b7ad412146c7b25e14491e397ff3 : SUCCESS in 15m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/019cac492c0541e98f0d54b6fd390268 : SUCCESS in 21m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00c80cb92cae4bf8b3455b369b984e3e : SUCCESS in 20m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/37ad82415f384a92a01766bf55bd24d3 : SUCCESS in 1h 06m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15ca98e2608e49f687755a2f35490c43 : FAILURE in 51m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ddacb0f7232470286c00ae01acbf9b8 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8ab6152ec8504f624fb961200119a5e5248198f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:44:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e39f7c0ae1254f6b8bfbc009ed32eab6 : FAILURE in 1h 23m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a3306c54f144924b210ccde78bccf0c : FAILURE in 29m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff00064b535a48a4af098db98f79b432 : FAILURE in 13m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cdd696a614445bfb3f5538ae8a7b4bf : FAILURE in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3ae08e46d150731080a35ffa26916a2110d1ab1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-17 13:15:50.000000000","message":"Patch Set 1:\n\nSeems to be about ~3M total for packages on CentOS. I think it\u0027s ok.","accounts_in_message":[],"_revision_number":1},{"id":"42a85358cfe4a74f50e411db9bd324d984ca66eb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-17 13:16:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a65ff1a5ae98e0456e372f330812bf1ee5c54fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 14:16:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ef083cbf07b50e3f8d1cef388123a1fc7f1511f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:17:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a9a59d8c1d68b5c844cb08ba34748c0a424b727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 15:43:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/50e7f30bf2044f8c9526e82fb08c2e9f : SUCCESS in 49m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/750b6aeddb1149f1a2f7ee5093483cf8 : SUCCESS in 1h 03m 24s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c6291695d1e145c29175be6fd86d6edb : SUCCESS in 1h 07m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b9ed75156747ceb140ba2c138c3190 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d23b3ae83c6483b84e2ef50ef8f53c6 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808af34f42a24d07bd02fca4747cd16e : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520199159d844056a4516920c225a9d5 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c62760d9ac4b018e9ce0c88ffb414a : SUCCESS in 25m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74483e7f4ea6499bae15337704432a77 : SUCCESS in 25m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/934710923f9d4af1b19c61a204bbf1fc : POST_FAILURE in 1h 19m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/823462094b7c4c9ba7345abb5f2917ac : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5e9543a675c551986477b20b9b250a7f66a1f6b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 16:36:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"876718afb09f966c0428d17a4797c1ebca971983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:32:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e36a18cff4e45008db4aed93e624064 : RETRY_LIMIT in 2m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b770374306bd4ed799781a9e959c3162 : RETRY_LIMIT in 4m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90065015b90e4c388350c7955fdc3edd : RETRY_LIMIT in 2m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9ca0624548240aab55aad6ff0fefce0 : RETRY_LIMIT in 2m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae259574d6c42a6b1c009bdd3a222666dd174ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:44:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e89b0b69500f45308bf64936e1a756c1 : SUCCESS in 28m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3903415bff4740a4bce1ff8c0baccc5f : SUCCESS in 28m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/295383aa80634debb0b1a2fa1af94faf : SUCCESS in 1h 05m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7e2feaad5741432abfe2ab8179098f2d : SUCCESS in 1h 00m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/14716ddfec71479e8b91515ca2027f01 : SUCCESS in 1h 03m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f24d2324e94465292ce67d5d69a8191 : SUCCESS in 17m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/48755a721cd1423fabb7b48d7da069c6 : SUCCESS in 18m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5536ac4922b64d7ba05755a2d054a4cc : SUCCESS in 47m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1452597267e4af88d89c99f3e76fd0e : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f46860adb54fde8329aed42b8c7058 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9228487535e04645af9e3a3aa90eb93a : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7762a52062c7424987a1c58f05affe23 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5e76515d314602b9cf5727c74b75ee : SUCCESS in 17m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1a02896ad2c4b6d9df67ffb1e1dc212 : SUCCESS in 23m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2aa436b55c7342f19bfb03c77c9c9c12 : SUCCESS in 21m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d595f73855f432cb81055ecdfdf416a : SUCCESS in 53m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e798e1ca5575459e91a3a812e6004dd2 : FAILURE in 40m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b733f3bfebd454bbeb75d58ab42416d : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fb4bcd4e92099b84100029f404dcd0da0e771bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:44:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"522b579dcefb989620927ceba6ce8d0fd2a26140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:57:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e827a8a78a60b3ea37d5304e2c0864661ee2dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:57:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/01d31f0460eb431bb114ee80f3400f8e : SUCCESS in 23m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bbc6cc5e3bfd4c30b29e94df444fa679 : SUCCESS in 1h 03m 10s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d0b2dc84687b4feb86223fcdc8d47e6c : SUCCESS in 1h 08m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a12ea6547a041d9b599503179969a03 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b2bedc0299468684ab832ddd1e86ff : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96ed97fc1dc64155888ddc2faec0e414 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd50ee9a7304dfbb6190cdfaeb1dba5 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b96b5017adb4d978f5861ebe5887583 : SUCCESS in 16m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f9ae93e28454583a6aaa9b108202757 : SUCCESS in 22m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08380b3a04b049908a6257c199cc1e99 : SUCCESS in 52m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff4423e3e4a04f3fb9603975233f6633 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"bb05dd8d7962fac3bcac36e809d34ab567f77f35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38477e93367f4fc989ee58be97e09337 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb71138fc61b7f7d1e214af2cb7327ddf87ea2ad","revisions":{"bb71138fc61b7f7d1e214af2cb7327ddf87ea2ad":{"kind":"REWORK","_number":1,"created":"2020-09-17 11:13:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/752434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/752434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/752434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/752434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/752434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/752434/1"}}},"commit":{"parents":[{"commit":"9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b","subject":"Merge \"ironic-pxe: RHEL/CentOS: install AArch64 modules on all GRUB architectures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a9ae6e295301cc8e60f81e62ec8fc9fd12c852b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-17 11:13:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-17 11:13:39.000000000","tz":120},"subject":"Add openvswitch packages to kolla-toolbox","message":"Add openvswitch packages to kolla-toolbox\n\nRequired by https://review.opendev.org/#/c/750631/\n\nChange-Id: I9e4e408ebc197b12ef128ec629fb648246ef6582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb71138fc61b7f7d1e214af2cb7327ddf87ea2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb71138fc61b7f7d1e214af2cb7327ddf87ea2ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
