)]}'
{"id":"openstack%2Fkolla~752512","triplet_id":"openstack%2Fkolla~master~I76ab6786b271963252278fcf12f1a28cab520fdb","project":"openstack/kolla","branch":"master","topic":"kolla-kall-for-better-docs","hashtags":[],"change_id":"I76ab6786b271963252278fcf12f1a28cab520fdb","subject":"Improve docs","status":"MERGED","created":"2020-09-17 16:38:45.000000000","updated":"2021-03-26 10:53:02.000000000","submitted":"2021-03-26 10:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"752512-kolla-kall-for-better-docs","meta_rev_id":"90bb5cbae460557703f780ab3bdd0741808eff0d","_number":752512,"virtual_id_number":752512,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 10:40:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-26 09:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-22 11:50:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-26 09:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-03-17 15:13:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 17:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-21 08:56:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-26 09:38:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c1f12e38f0dd356c5f271776b003c4a12ce81e64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-17 16:38:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b7e8ad3ac66ae573337ac862bdca27aec98b7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:08:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623fc3f96bb44612bd9dc86f44a57143 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fa2f3145254c58b065426489e19061 : SUCCESS in 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fb7d17b4fd8bb7efa51ec177b8c32186a9f9cf36","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 08:56:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"84f9df7c9bfb8a8c2c03d547e2bdf088e0501a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-20 21:12:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f2724060e32b6576e8a9ef430c6e09fc036330ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-20 21:16:06.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nI have fixed the merge conflicts and added a bunch of new changes. Please let me know if you are happy with these and let\u0027s merge.","accounts_in_message":[],"_revision_number":2},{"id":"e9775638e810543e87cdf3faa19e52ab779a21e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 21:24:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0775add67c0488284aa596602705d83 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c81ed462c84a628ac1cfbfb9d45e9b : FAILURE in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"21d9a18bc3449be3a41259cc60d4e4782e8a1ac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-21 09:16:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9337db8cf972feb82827df8117d80402e5814eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 09:27:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd1b6457b0f49468b5a9a275744c631 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894b4f6775be41468833547990142fb0 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bc34fdd86fd9ab7efbbc2abf7ee6ee15c284b485","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-03 10:45:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"135a952cae0f9e631287990b9e0b3a3d3ca75bc6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-15 12:36:48.000000000","message":"Patch Set 3:\n\n(7 comments)\n\nFeels like I am doing a second PhD in these docs... 😊","accounts_in_message":[],"_revision_number":3},{"id":"360f5e42390f5d4d713646db3d545331ac189c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 09:46:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c534c9b23b4abe04d82ff9f6d96c06759f52ff22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:54:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427b2b16d4b041b4908c89328681a758 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef179622b5a4331a6aef0c175786973 : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"62f3c9573fb08ba682ca3046ae71297654d5d08e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 10:14:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8eb4fbeef7dd805569490b1af2aa1d3bd92475e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 10:29:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2577cb1a1f2f43cfaf0f893085cfec9f : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c434ed34ec3400087e5abacd8558d92 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":5},{"id":"5047b9908601a7a4349a80b48e835009eef5e84a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-17 15:13:17.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"8ff36260ad8d976109625e51c00fa3b63a3265f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 11:50:05.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)\n\nThanks for completing your dissertation","accounts_in_message":[],"_revision_number":5},{"id":"3c80136a7fe2852421479ebc331fcfbef79d0aa1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:38:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c8ead32dd9a0c9a697a0529b94a3f3ccb2c233c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 09:39:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8e73001a72f49114ff07a40423a47d56d17b3e01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:40:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43143bbdbbe41b8a956b76b6ff7ee3a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2490de60da4f6e86c3c5d959709003 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":5},{"id":"1ef6915437dca51753830263064ae99ad8fb6b7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:40:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"90bb5cbae460557703f780ab3bdd0741808eff0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:53:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc899673cd7a4e2ca9445cbc459fb8b8 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d9efe9cac63e6916d1012e39ec0912e9693e7c52","revisions":{"67bac6752fc6dd2fdd9e010c954d53200cafb94c":{"kind":"REWORK","_number":1,"created":"2020-09-17 16:38:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/752512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/752512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/752512/1"}}},"commit":{"parents":[{"commit":"0abf8bdac90646f8342f358514461978786d1fb9","subject":"Merge \"[CI] Support building source images with in-review changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0abf8bdac90646f8342f358514461978786d1fb9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"subject":"Improve docs","message":"Improve docs\n\nPart of changes discussed during Kolla Kalls.\nThese make our documentaion easier to digest.\n\nChange-Id: I76ab6786b271963252278fcf12f1a28cab520fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67bac6752fc6dd2fdd9e010c954d53200cafb94c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67bac6752fc6dd2fdd9e010c954d53200cafb94c"}]},"branch":"refs/heads/master"},"78f5b573a2579640c376ad489530a80275adb940":{"kind":"REWORK","_number":2,"created":"2021-02-20 21:12:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/752512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/752512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/752512/2"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-20 21:12:34.000000000","tz":60},"subject":"Improve docs","message":"Improve docs\n\nPart of changes discussed during Kolla Kalls.\nThese make our documentation easier to digest.\n\nChange-Id: I76ab6786b271963252278fcf12f1a28cab520fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f5b573a2579640c376ad489530a80275adb940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f5b573a2579640c376ad489530a80275adb940"}]},"branch":"refs/heads/master"},"e2ab13060060e5edc0dc69216d964e503a2ccc5a":{"kind":"REWORK","_number":3,"created":"2021-02-21 09:16:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/752512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/752512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/752512/3"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-21 09:15:26.000000000","tz":60},"subject":"Improve docs","message":"Improve docs\n\nPart of changes discussed during Kolla Kalls.\nThese make our documentation easier to digest.\n\nChanges in conf.py are to avoid warnings from sphinx.\n\nChange-Id: I76ab6786b271963252278fcf12f1a28cab520fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2ab13060060e5edc0dc69216d964e503a2ccc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2ab13060060e5edc0dc69216d964e503a2ccc5a"}]},"branch":"refs/heads/master"},"6cb8b04c23299880e6068d9cb36cf1defe6c38e7":{"kind":"REWORK","_number":4,"created":"2021-03-17 09:46:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/752512/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/752512/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/752512/4"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 09:46:00.000000000","tz":0},"subject":"Improve docs","message":"Improve docs\n\nPart of changes discussed during Kolla Kalls.\nThese make our documentation easier to digest.\n\nChanges in conf.py are to avoid warnings from sphinx.\n\nChange-Id: I76ab6786b271963252278fcf12f1a28cab520fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6cb8b04c23299880e6068d9cb36cf1defe6c38e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6cb8b04c23299880e6068d9cb36cf1defe6c38e7"}]},"branch":"refs/heads/master"},"d9efe9cac63e6916d1012e39ec0912e9693e7c52":{"kind":"REWORK","_number":5,"created":"2021-03-17 10:14:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/752512/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/752512/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/752512/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/752512/5"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-17 16:36:47.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 10:14:51.000000000","tz":0},"subject":"Improve docs","message":"Improve docs\n\nPart of changes discussed during Kolla Kalls.\nThese make our documentation easier to digest.\n\nChanges in conf.py are to avoid warnings from sphinx.\n\nChange-Id: I76ab6786b271963252278fcf12f1a28cab520fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9efe9cac63e6916d1012e39ec0912e9693e7c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9efe9cac63e6916d1012e39ec0912e9693e7c52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
