)]}'
{"id":"openstack%2Fkolla~752616","triplet_id":"openstack%2Fkolla~master~Ie26c0eb42c128aa997b4255fcf7e7d0da015cb81","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie26c0eb42c128aa997b4255fcf7e7d0da015cb81","subject":"Add the missing doc/requirements into running-tests.rst","status":"MERGED","created":"2020-09-18 05:17:01.000000000","updated":"2020-09-18 07:46:41.000000000","submitted":"2020-09-18 07:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752616-1600415052209-515a9e8b","meta_rev_id":"a05d0f56c60253e8899865900b1fc169481a49e9","_number":752616,"virtual_id_number":752616,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 07:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-09-18 07:19:07.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-09-18 07:19:07.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 07:19:07.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-09-18 07:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6de319168b6530bd47cfc5614dae315736fbe47d","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-18 05:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eddf7d668178907302cbcd4c94d482e130e05084","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-18 05:19:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d983d6752c8ac1d788fc0479d0d099cca91f01a1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-09-18 05:20:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9a6914610c34ed57bb837abea6c3c3a48aa6674","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-09-18 05:20:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b7e83c08e0e6cdd8507a5197dac3f0fecfc87df","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-18 05:21:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7d30f82d2ec5f61e5f99117bd6f14b3fc2148ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 05:41:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd68fe5714449f5ada3b7e4583aff67 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1129664fd2974849b203d26d10493938 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":3},{"id":"58c0e0bf2f58230280efdaf1a0c9abf0801d46aa","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-09-18 07:19:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nTiny docs changes, merged fast","accounts_in_message":[],"_revision_number":3},{"id":"e3aaf3341bc44f41ee8e67a026502f56cbcf9863","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:19:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c976732ece58e0326e36555b64503b32b7e96154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:44:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e5fe08f217460a9b584d3aed813dcd : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a9fe77b23f4c16ae983dec2bd96a1b : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":3},{"id":"17b87b525441cd3e15a741e7d7d50181e52a9832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:44:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a05d0f56c60253e8899865900b1fc169481a49e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:46:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5d28ee3ce6485dbc52cef2697d57c0 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1cf163191fbcd3b68e540bb37cc0e9a8a92ef223","revisions":{"0376b06584b06464940c60f4d45b16e7f3b215c0":{"kind":"REWORK","_number":1,"created":"2020-09-18 05:17:01.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/16/752616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/752616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/752616/1"}}},"commit":{"parents":[{"commit":"88ac478d87538798e4cb7c9e54b2069b7fc58afb","subject":"Merge \"Update py37 to py38 for running-tests.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88ac478d87538798e4cb7c9e54b2069b7fc58afb"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:15:07.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:15:07.000000000","tz":480},"subject":"Add the missing doc/requirements into running-tests.rst","message":"Add the missing doc/requirements into running-tests.rst\n\ntox command install dependencies from the ``requirements.txt``,\n``test-requirements.txt`` and ``doc/requirements.txt`` files\n\nChange-Id: Ie26c0eb42c128aa997b4255fcf7e7d0da015cb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0376b06584b06464940c60f4d45b16e7f3b215c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0376b06584b06464940c60f4d45b16e7f3b215c0"}]},"branch":"refs/heads/master"},"9afae7a2447312dc14d9dec92c098c1116e41e7c":{"kind":"REWORK","_number":2,"created":"2020-09-18 05:19:04.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/16/752616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/752616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/752616/2"}}},"commit":{"parents":[{"commit":"88ac478d87538798e4cb7c9e54b2069b7fc58afb","subject":"Merge \"Update py37 to py38 for running-tests.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88ac478d87538798e4cb7c9e54b2069b7fc58afb"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:15:07.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:18:32.000000000","tz":480},"subject":"Add the missing doc/requirements into running-tests.rst","message":"Add the missing doc/requirements into running-tests.rst\n\ntox command install dependencies from the ``requirements.txt``,\n``test-requirements.txt`` and ``doc/requirements.txt`` files\n\nChange-Id: Ie26c0eb42c128aa997b4255fcf7e7d0da015cb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afae7a2447312dc14d9dec92c098c1116e41e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afae7a2447312dc14d9dec92c098c1116e41e7c"}]},"branch":"refs/heads/master"},"1cf163191fbcd3b68e540bb37cc0e9a8a92ef223":{"kind":"REWORK","_number":3,"created":"2020-09-18 05:21:43.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/16/752616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/752616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/752616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/752616/3"}}},"commit":{"parents":[{"commit":"88ac478d87538798e4cb7c9e54b2069b7fc58afb","subject":"Merge \"Update py37 to py38 for running-tests.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88ac478d87538798e4cb7c9e54b2069b7fc58afb"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:15:07.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-18 05:21:36.000000000","tz":480},"subject":"Add the missing doc/requirements into running-tests.rst","message":"Add the missing doc/requirements into running-tests.rst\n\ntox command install dependencies from the ``requirements.txt``,\n``test-requirements.txt`` and ``doc/requirements.txt`` files\n\nChange-Id: Ie26c0eb42c128aa997b4255fcf7e7d0da015cb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cf163191fbcd3b68e540bb37cc0e9a8a92ef223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cf163191fbcd3b68e540bb37cc0e9a8a92ef223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
