)]}'
{"id":"openstack%2Fkolla~753944","triplet_id":"openstack%2Fkolla~master~I974dfba1b3f939bb5d56e994c080da5673c7953b","project":"openstack/kolla","branch":"master","topic":"ubuntu_binary_heat","hashtags":[],"change_id":"I974dfba1b3f939bb5d56e994c080da5673c7953b","subject":"Add missing heat deps for deb-binary","status":"MERGED","created":"2020-09-24 05:38:51.000000000","updated":"2020-09-28 11:33:15.000000000","submitted":"2020-09-28 11:30:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753944-1601292655228-6ef20142","meta_rev_id":"00e1054a5306717ac10a5838d14995dac88586d2","_number":753944,"virtual_id_number":753944,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 11:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-28 09:36:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-26 09:15:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-28 09:36:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 05:50:17.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-26 09:15:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-28 09:36:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-28 11:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d487f8d0c40431fc384ee376363858ba78da426","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-24 05:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfbd1079e4de23f8d084443bc8f584a3b5677c36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f11f3bbf04964c1aafa1cc65e851a82d : FAILURE in 50m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d14257b0c09d437fa038a1ad1dc5c993 : FAILURE in 30m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a32bf7ba44d849aaaecfa7f93caae0b1 : FAILURE in 56m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0556cc9dd76448491e84d5b86354f8c : FAILURE in 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8efdd03a0dd2572fe6e24ea26ee883891b8e714","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-24 07:27:16.000000000","message":"Patch Set 1: Code-Review-1\n\nplease add revert comment","accounts_in_message":[],"_revision_number":1},{"id":"9ec94eaa8c089316192249b40b2a8e3769c063ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 07:48:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/493f3630e60b48729623bbd6a2daf1ff : SUCCESS in 19m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/692109d49ceb427da914c19458163c09 : FAILURE in 1h 21m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/91dfda4b39344c4fbe29c0f505df6227 : SUCCESS in 1h 30m 36s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d941ef4096054ad3885d9f7388d717b9 : SUCCESS in 1h 30m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/41cb3f4285bd40d1aee38e0e46b24118 : SUCCESS in 1h 37m 12s\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f3d3847001754d07ae4e9b74df5d216b : TIMED_OUT in 2h 06m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d4c010599154ed2bfa28f443179a641 : SUCCESS in 56m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/719173485ad04ba2be0ffb625c969285 : FAILURE in 1h 00m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a710b48e46b244409adfbcb4cf6aa027 : FAILURE in 1h 04m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb565248277d4e3ab1aa5cd2a860a20d : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d95097d34fb4198969bd2406764f8b2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dd90cf060da44e3bfd97c832fa0a478 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cc7b2cca0654efa9ac6448af54615d4 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feba730984154a1189cf90f6dd461d6b : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb6ba2b2411443f3ac241c37c45c7585 : FAILURE in 1h 05m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be32820174d04f6a9abb70e5d4586c49 : FAILURE in 1h 04m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0c2ef09f44b46d0a1303b41ec5bd1b5 : SUCCESS in 1h 25m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23a8b68902224fcbba647379b821f526 : FAILURE in 45m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb4dc5f39c384e689972fb8489443eea : SUCCESS in 1h 39m 37s","accounts_in_message":[],"_revision_number":1},{"id":"aef66507f0a76cbf1948dba16f2880b81add6886","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 08:04:37.000000000","message":"Patch Set 1:\n\nu-b still failing :(","accounts_in_message":[],"_revision_number":1},{"id":"1def850b0c970b996d13748919737c1d67d13e02","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-24 13:52:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7a974ceb7c450c04db47fa98f37d7dcae19a669","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:01:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/724befee54bf4f3d969e4b3f3493f8ec : FAILURE in 42m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58636b9f94f64a64a7865202f66ea617 : FAILURE in 32m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fab4f8d804854c9ab89e22f1b4a6626b : FAILURE in 53m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/08fa6caa22ed436291344117c2bc93df : FAILURE in 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32417d8fddd5cd613f7e1674a3394be91c2e8867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:14:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c715cc335c8b464e83bee262fe60c05e : SUCCESS in 19m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ba9c9d386b14c29afa1b10ec87d622b : SUCCESS in 24m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/82c0daa9051544798591aa385d75ae12 : SUCCESS in 1h 04m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0edff734155447edaeb565fc0b722b29 : SUCCESS in 1h 10m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/00add2bc134040bba314d7dd96800c19 : SUCCESS in 1h 17m 49s\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8d606440a7a3499badfd8f312a8adaba : TIMED_OUT in 2h 07m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3eff0e4cf3c04d77b281a1aaa095ec5b : SUCCESS in 1h 21m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/236c7bb56e2d45c5a654c5224d761a5b : SUCCESS in 26m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/88c1c6ef966e405ba1b89f2a29baef29 : FAILURE in 36m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53b97327a3d641a693b6dc803fcb6ca1 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2476185e274bceb0ab239380390fc9 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be7221dcd77c4ac3ba664f0ecddc0e3d : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ead87baa804d1ab790e05e9eecbe0d : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb5acd1cee54682b5e10c9f4964a175 : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd0c1d9eacf3400781cf5b1ba635d252 : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/58acb28b5eb74e6aaa0854d83c9cbd32 : SUCCESS in 31m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ba482010aab4926872378bcefee9361 : SUCCESS in 57m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f11c4ca3517c4f7d8a33ee4c87703029 : TIMED_OUT in 2h 03m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/593da3516043442ea372a372bbc7b733 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c4d529bd7bf7874d3807a8866b496318bad1eecb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 16:22:09.000000000","message":"Patch Set 2:\n\ntimed out","accounts_in_message":[],"_revision_number":2},{"id":"1222148e1fa38dc14274094969e7e96257197ace","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-24 16:40:59.000000000","message":"Patch Set 2: Code-Review-1\n\nplease add revert comment (I feel like repeating myself here)","accounts_in_message":[],"_revision_number":2},{"id":"cb9adad9c48e778ae0945b5f4ec099b3e560c15a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-25 05:45:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a545fa9b39703f1f8d79846d3172f95833b3e422","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-25 05:50:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a44be6263a6131baba648b76632124dafcd39f23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:00:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c8ffdc9b1ab4e0b92e2872e53bef93e : FAILURE in 7m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6aa4e911f1144318cc787ec086fa483 : FAILURE in 3m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1311dfd51f45472283ce68d4a00909ce : FAILURE in 9m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c758e0a561a4170af0673aeb5fd3459 : FAILURE in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d673371bf7d50c41dd389e4935b9b0c5d545c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:06:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/984a6cf8a6fc4d71ac2c12315c656dfd : FAILURE in 4m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ddfe16bfcf574739a93840afad12838b : FAILURE in 4m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6f56051e74140e7be76dc719269473c : FAILURE in 9m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/20bcbfb9d9a048588254883e9c9dc47f : FAILURE in 9m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f526c46a6b9c4c6db0c66d5b573ea187 : FAILURE in 8m 57s\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ea549b41d7e0479f82cfb1f6a0a4c493 : FAILURE in 11m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/94ba2b2ebe5348cfac71ce9a1aab298c : FAILURE in 3m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f08a83802b142eea4e9c303673c62f7 : FAILURE in 3m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/88eb0e4bcd1b42f8aa9b9d9dfb391e71 : FAILURE in 7m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60a3e203c94046efa7663144f37dc29e : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc7bc7218bd4d0a8cf97c15493ac04e : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71fb1140be5b4cd3b2eec9e091e54b22 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/009aea82beb6412494fc227711c90db9 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fb00a052464721a473e1499ebccdd5 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f24d2020304f43799c75d77cae1671fe : FAILURE in 3m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/31a4425602124a42bce5f2e48d2988c9 : FAILURE in 4m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d3766d19a534d14a90f11348fd69e5f : FAILURE in 8m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ceab3cfa9c8433b86d9044245fab944 : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0fe4ab3f89e741fd8f1e75e367706245 : FAILURE in 19m 34s","accounts_in_message":[],"_revision_number":3},{"id":"cd51dc8644cd69df6ff79bbf31d7b7c8f08ee656","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-26 08:56:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3680fc0b788acd8318850c368a03c081603a8126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 09:10:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7b929a5094bc4fb38d543bd3ad49e121 : FAILURE in 4m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fc1627b90b0f4def84ac99df541db206 : FAILURE in 4m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/229477cc03334a6ba83668e100064b06 : FAILURE in 8m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f6816ea847964199a091ebfb1760ed88 : FAILURE in 10m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7901a31836b2469a960cb311e08d853a : FAILURE in 9m 33s\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/edc0ef8909874e1f86a5101be764213e : FAILURE in 10m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d53e338fb39c475a9fc3d2a54270c2ff : FAILURE in 3m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9f358a2f5c4142c581567828b6b7e6a7 : FAILURE in 4m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0dd2e6b3a93a48bebec6040df238a022 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27c38bf81bbf4a7893b2219446acb5d1 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3fc13d0541c49f88c0414d2a25bffdc : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14fca488613042869d7996a16dd177cf : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db950aeafa554df5900f1fa0bf24f168 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35edb0f0f544faea7d6655f9a32d0df : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71b8a13d47f64ccbb93b4bf2015bc573 : FAILURE in 3m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14b91eae392942f596a3ce4bd2c34688 : FAILURE in 3m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b3d407dde8c46de9a25a25293e4d77c : FAILURE in 8m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8552c6a84d5640edb64a60dbddd96c36 : POST_FAILURE in 7m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea0ec304770c42c2ad1baffc10313b57 : FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":4},{"id":"cc6b1442c795f68aa20088504c4f106b5fc30458","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-26 09:15:29.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2e8d9390f6746290d895f19521ec4a3b7b5b5c69","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-26 09:15:54.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2cd3db754d729347982ee30a3c22db083790b2c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 10:19:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9582bd2c35f34d2c9240fbc57b56e887 : SUCCESS in 42m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e94c0d02afa44460bbc33dc4e281436f : SUCCESS in 29m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2477c4f979b14854938b12d09b493bfd : FAILURE in 52m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6efadec7051448ae8f53cf039673c062 : FAILURE in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e8e053dbc8258505c3f5832ddf49618a12c87937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 11:24:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5af39815c7ed4e48a3f23b1778239a78 : SUCCESS in 25m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7a67e17fe3a4bd7adbe7f00df6c6993 : SUCCESS in 26m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c94551646b6c444b94b33f9627f2ba25 : SUCCESS in 1h 03m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4ff702e1b24644508cbe3c3a69f70f01 : SUCCESS in 1h 01m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c39fb3c144c245978d289f21db50b49a : SUCCESS in 1h 09m 14s\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/96275e0e099e46768cb1a29886c2ff79 : TIMED_OUT in 2h 05m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e777ad9ca45f41c7908f6a6d1ff8063c : SUCCESS in 18m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a0bea6e32494f629eaee1f42399dbec : SUCCESS in 21m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a623bc1aedb543a587d080e809c5a4fd : FAILURE in 35m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f732ebbf167425e9240675f278dfc7c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049d6e0d926a481d830c42a54e6d92cd : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b770b88f77e8470eb055f4d5e014ee5c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/720a9fcb256842beab5740d5b87c386e : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3204cd1aaa44cf0a821b288e45fe30b : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d99fbcb817fd4f00af894b21387f7051 : SUCCESS in 25m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2ce53d289c84a469e02c4125ce283db : SUCCESS in 18m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48888785ce274b8f8af4a37a3d6e8091 : SUCCESS in 52m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d1272c389324bdf8781947da8e129c3 : SUCCESS in 54m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ab43aaa4010439fb906ecd082a01521 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":5},{"id":"b09c4f6e87dc4c13f4eba41d64b8dbaa1dad0e33","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-28 09:36:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"cfaa8d8b01a1e4f2f13b01b76e150064f32db06c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 09:37:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f03d0244b91ef0f332d6f29600a4054f76569d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 11:30:54.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b2bc4844efd04b0a81bc1763a0e714f0 : SUCCESS in 26m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c1ad29cdacb4434298760106f367dc6d : SUCCESS in 1h 03m 52s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b104563aa1ef4efab26accec42d80a66 : SUCCESS in 1h 08m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f130f811c8aa440cb014be4ce38b819d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430f05617ee5464ea0dce4ca0b646769 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3b837034fc49dcac94c1a296416282 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45105180bbf94e6ca470e1f380520867 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9128b90f8cbf49528c3efbc02b0b640e : SUCCESS in 6m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3189b09b7914426989ad8c0b9c5ce8cc : SUCCESS in 18m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39f024ba8e32413c9a729994606f725e : SUCCESS in 44m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5783df57b8e24648b3f3fe29fee9502f : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":5},{"id":"2a7ed47cd0e0f99f6dab6274721b16fb9a7be5f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 11:30:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"630577af41c3764d1ad27c833fa718fd10320a60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 11:33:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd1dee8d32b46c8b420e70c730d9f6f : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d39c3e4a85ff4f3bce028c5a79bccf783512772f","revisions":{"06ffd4ba37acc66d5862600cd501cbddc8e89631":{"kind":"REWORK","_number":1,"created":"2020-09-24 05:38:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/753944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/753944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/753944/1"}}},"commit":{"parents":[{"commit":"f2daa5fa85216e302e171c288f6a86cd01fe61c2","subject":"Merge \"Fix typo and capitalisation in help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2daa5fa85216e302e171c288f6a86cd01fe61c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:35:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:38:48.000000000","tz":120},"subject":"Add missing heat deps for deb-binary","message":"Add missing heat deps for deb-binary\n\nheat-engine in V relies on vitrage and zun python client libraries, there is\na bug [1] in progress, but let\u0027s fix it faster in Kolla - and revert once\nUCA fixes it on their side.\n\n[1]: https://bugs.launchpad.net/cloud-archive/+bug/1893935\n\nChange-Id: I974dfba1b3f939bb5d56e994c080da5673c7953b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ffd4ba37acc66d5862600cd501cbddc8e89631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ffd4ba37acc66d5862600cd501cbddc8e89631"}]},"branch":"refs/heads/master"},"873e78a24171dd7120b27420a00560dcd4dd9caa":{"kind":"REWORK","_number":2,"created":"2020-09-24 13:52:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/753944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/753944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/753944/2"}}},"commit":{"parents":[{"commit":"f2daa5fa85216e302e171c288f6a86cd01fe61c2","subject":"Merge \"Fix typo and capitalisation in help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2daa5fa85216e302e171c288f6a86cd01fe61c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:35:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 13:52:19.000000000","tz":120},"subject":"Add missing heat deps for deb-binary","message":"Add missing heat deps for deb-binary\n\nheat-engine in V relies on vitrage and zun python client libraries, there is\na bug [1] in progress, but let\u0027s fix it faster in Kolla - and revert once\nUCA fixes it on their side.\n\n[1]: https://bugs.launchpad.net/cloud-archive/+bug/1893935\n\nChange-Id: I974dfba1b3f939bb5d56e994c080da5673c7953b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/873e78a24171dd7120b27420a00560dcd4dd9caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/873e78a24171dd7120b27420a00560dcd4dd9caa"}]},"branch":"refs/heads/master"},"ba0790de0d21e2d8809548eadb3c02938daff569":{"kind":"REWORK","_number":3,"created":"2020-09-25 05:45:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/753944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/753944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/753944/3"}}},"commit":{"parents":[{"commit":"f2daa5fa85216e302e171c288f6a86cd01fe61c2","subject":"Merge \"Fix typo and capitalisation in help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2daa5fa85216e302e171c288f6a86cd01fe61c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:35:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-25 05:42:52.000000000","tz":0},"subject":"Add missing heat deps for deb-binary","message":"Add missing heat deps for deb-binary\n\nheat-engine in V relies on vitrage and zun python client libraries, there is\na bug [1] in progress, but let\u0027s fix it faster in Kolla - and revert once\nUCA fixes it on their side.\n\n[1]: https://bugs.launchpad.net/cloud-archive/+bug/1893935\n\nChange-Id: I974dfba1b3f939bb5d56e994c080da5673c7953b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba0790de0d21e2d8809548eadb3c02938daff569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba0790de0d21e2d8809548eadb3c02938daff569"}]},"branch":"refs/heads/master"},"912d29773ca66d0eb95985591bf7bd1aafc98168":{"kind":"REWORK","_number":4,"created":"2020-09-26 08:56:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/753944/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/753944/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/753944/4"}}},"commit":{"parents":[{"commit":"f2daa5fa85216e302e171c288f6a86cd01fe61c2","subject":"Merge \"Fix typo and capitalisation in help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2daa5fa85216e302e171c288f6a86cd01fe61c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:35:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-26 08:56:15.000000000","tz":0},"subject":"Add missing heat deps for deb-binary","message":"Add missing heat deps for deb-binary\n\nheat-engine in V relies on vitrage and zun python client libraries, there is\na bug [1] in progress, but let\u0027s fix it faster in Kolla - and revert once\nUCA fixes it on their side.\n\n[1]: https://bugs.launchpad.net/cloud-archive/+bug/1893935\n\nChange-Id: I974dfba1b3f939bb5d56e994c080da5673c7953b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912d29773ca66d0eb95985591bf7bd1aafc98168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912d29773ca66d0eb95985591bf7bd1aafc98168"}]},"branch":"refs/heads/master"},"d39c3e4a85ff4f3bce028c5a79bccf783512772f":{"kind":"REWORK","_number":5,"created":"2020-09-26 09:15:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/753944/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/753944/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/753944/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/753944/5"}}},"commit":{"parents":[{"commit":"f2daa5fa85216e302e171c288f6a86cd01fe61c2","subject":"Merge \"Fix typo and capitalisation in help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2daa5fa85216e302e171c288f6a86cd01fe61c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-24 05:35:59.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-26 09:15:26.000000000","tz":0},"subject":"Add missing heat deps for deb-binary","message":"Add missing heat deps for deb-binary\n\nheat-engine in V relies on vitrage and zun python client libraries, there is\na bug [1] in progress, but let\u0027s fix it faster in Kolla - and revert once\nUCA fixes it on their side.\n\n[1]: https://bugs.launchpad.net/cloud-archive/+bug/1893935\n\nChange-Id: I974dfba1b3f939bb5d56e994c080da5673c7953b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d39c3e4a85ff4f3bce028c5a79bccf783512772f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d39c3e4a85ff4f3bce028c5a79bccf783512772f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
