)]}'
{"id":"openstack%2Fkolla~755346","triplet_id":"openstack%2Fkolla~master~Ib2bf51d4c6e2a574b98671a699777cd8a9a22768","project":"openstack/kolla","branch":"master","topic":"fix-neutron-migrations-order","hashtags":[],"change_id":"Ib2bf51d4c6e2a574b98671a699777cd8a9a22768","subject":"Fix Neutron migrations order","status":"MERGED","created":"2020-09-30 16:46:15.000000000","updated":"2020-10-01 20:03:00.000000000","submitted":"2020-10-01 20:01:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755346-1601582498928-f61217dc","meta_rev_id":"e1936ee26fb42251aa454505cf70ec676d0501ee","_number":755346,"virtual_id_number":755346,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 20:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-01 08:33:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-10-01 08:36:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-10-01 08:36:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2020-10-01 08:04:12.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-09-30 16:59:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 08:04:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-01 08:33:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-10-01 08:36:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-01 20:01:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0e6d072134a7fc0739fa1ffd76855fe8059ef3b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-30 16:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63fc46f13b3db030f700ad5163d57df921d6545d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-30 16:59:12.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"9ad1f7fd1d999d1cc81c153e84e4575f398c9812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 17:52:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/63583d9022b3407983ef877fcb0c8aa5 : SUCCESS in 52m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56427c2198a24dc699cc8c07bff6d323 : FAILURE in 41m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7673143127c4d2994ee81c942cded6c : FAILURE in 1h 04m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d406dba0bfd44b89d32bd15ec67607c : SUCCESS in 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbb8a62885367da9da18acee4c77705ca9269a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7a5d767de98f4b518441de72d27e50c7 : SUCCESS in 25m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/26bbefcbc6b24faca0db1f66a8124b43 : SUCCESS in 28m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b08c0644e11d4e09a037132ebf533f0f : SUCCESS in 1h 04m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ad2c8f492b2646a489c24b31ef18b90d : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0da2f6162a634c958577bd0e6c3e93cd : SUCCESS in 1h 20m 39s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/285034c3265f48b18efa5d45c72708a9 : SUCCESS in 1h 18m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b9847f61cb084b8fafa4065207efeed7 : SUCCESS in 18m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f547e3a631cc431bbc0fdaecbd6af3a3 : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa0053acd3e64d1db115bdf33b20e230 : FAILURE in 43m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a025c9fd10ce413a9c250798c7664afd : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a453cdc9f3472691eb79d89e5c1322 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/264658a89ca3416494881d038777b1d5 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b864e33ef66e4ef09b367cdb63e64844 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb16b3aa055d4e8d80836e644d308b37 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/319e6a6c0fdc475b8684b2632a1b4266 : SUCCESS in 21m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30f6dbdd7981484b839302ee3073c574 : SUCCESS in 23m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9bd45c9700c84a25b13e4f02327d2a3b : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb219145376f4eab9f44a955dcbbcb39 : SUCCESS in 49m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/98375b8021ad481cb06de974fea1b11e : SUCCESS in 1h 02m 28s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/5a50e9d72efb4468abb569ed8615d359 : SUCCESS in 52m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/9c9fb14fb54a4913b98b69523f8e2804 : SUCCESS in 1h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1584aa47221fe5e2a201c632cfc044b5e32c263","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-10-01 08:33:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0badeb581911e0cc887d5aa1e887a225553ac137","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 08:36:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ac0c6490005f747df7464e1bce42221f8d6f783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:37:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06d6eecc620ebe89a7ad1bc6016052c43d9f089f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:46:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4d6a6502cd184efaac233ac1074a0cf5 : SUCCESS in 44m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94c925041119498a8e3fef6c506f21a1 : SUCCESS in 1h 00m 47s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0846adece6bb4e5da73f0562312bdbc4 : SUCCESS in 1h 08m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ef9d370fb554283a4b3b1ec88a583cd : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbca33572d44b36818e7f19c9e3d113 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5fe47b5f8eb44bfaf1c3bb797f9ec15 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e12235dc56ad4cddbe4f5c02e8c569c0 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5444ccd50bad42028383f3a408721923 : SUCCESS in 7m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f939965eaecd44cf8010ea6212f024ba : SUCCESS in 20m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dde33353b0974d5d8f4d252ec1558358 : FAILURE in 46m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60435ecec0dc4336873ff32127b7e57e : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"39aba233f289f5ce112310574a21cd08f7eb89f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-01 11:11:11.000000000","message":"Patch Set 1:\n\nrecheck\n\nour favourite, although rare, \"keepalived decided you don\u0027t need a VIP address anywhere\"","accounts_in_message":[],"_revision_number":1},{"id":"e2f836ac4dd940a659bd67fef83f1525a6944257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:04:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9c6285d17be4882a8b77b5c42e2f58b : FAILURE in 51m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c0d723d8dd146fe84d9f9c7213fd3cc : FAILURE in 30m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/662aa3b7782a472e80d2429ef973af2e : FAILURE in 25m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc82cde147c640fab7ee4ab46435bfd0 : FAILURE in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a94270481463dfe062661c49126e25291008c889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 13:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9b865eba5e8b47e286d5eebaeda8ced8 : SUCCESS in 29m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f5e5fc95f496463790eea6866ef2b807 : SUCCESS in 25m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5ddd995813ef47089f9e50bb261f345f : SUCCESS in 1h 02m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8a17610d353d4df8ace148a775cc1318 : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/55f8665b10294b89b276e893beb0c0dd : SUCCESS in 1h 07m 01s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/dc5a560daf244929966001cecc5e5347 : SUCCESS in 1h 25m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/24298ad416314f00b52a5885f158958e : SUCCESS in 17m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62636ee22aab448fac63689d8a9e98e3 : SUCCESS in 21m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d92584aeeba4719b067fdf04109c388 : SUCCESS in 1h 03m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a38f89979174f3393ae1e15b28962e2 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef9f5ee1a804e7bb873013c09aff07f : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efb774bd6c154091848fcee571d1755e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5560a9fb99dc4f11a4f712a73c09c2ab : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7377ebd3f1141069e6b341c8cbdbc92 : SUCCESS in 9m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/567ec5a09e43487abd0688828fad3167 : SUCCESS in 20m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/936e9a302f814c248a969fbdc5743211 : SUCCESS in 21m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8c835c906014182bb273290fbc2996d : SUCCESS in 49m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d2e60b4a148405ca114d21847ba9e9e : SUCCESS in 50m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/880db963309548e38458db1e4d9fcf1b : SUCCESS in 1h 05m 53s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/be64b6946a474782a3b909af0471d0d8 : SUCCESS in 45m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/dcebf926e3bd43689cba9644c3880d98 : TIMED_OUT in 2h 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7d1057d70f67e4339a08f22c2682cf2d8789e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 13:36:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7f7faaee32e27305879fd4e1646ef9faef888e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:14:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b47446d572d445f19601ea3c76fc9640 : SUCCESS in 23m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/92cee5b11c824172a86076d3a073b8b0 : SUCCESS in 53m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/78282a586528414da61cf0ea46cdaf70 : TIMED_OUT in 2h 35m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c15f2fb5d3d9407983aae96fb85dc459 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa2148ff34243bdb0560a70bdd4eba6 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50c84ea170f54e10a412196beb9b9593 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1136549092d34cdcbebacd4943ad3a3f : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4be77eeeb54c9fb12a88f79943fd9d : SUCCESS in 5m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/13ce1a2c75dd4683912d9422c51c8009 : SUCCESS in 20m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/002cc6a672de4c279ee377dc9608c7fe : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c40a6c77b1dd4b798eca8201eb771b84 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5c43158afef895a9ac4e7f7561b67857b2ff17a8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-01 16:19:35.000000000","message":"Patch Set 1:\n\nrecheck\n\n\u003cput lots of favourite swearing words in here\u003e","accounts_in_message":[],"_revision_number":1},{"id":"035ab06d834ee24e191c787748a0c189810266fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:26:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/405fb3ab36f0485a8bb538f6348b8964 : FAILURE in 46m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c67780cd22b488e9b2b2badfdcff64f : FAILURE in 29m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/377553d54a70438ca61e446f9085a370 : FAILURE in 29m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e3e022d0d314940aeadf5e8a4a5499b : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"674e6e2df24cc1e1ae89879110507aa4960e0f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 18:45:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/344cd22af0424e8194d3326bcc3ed825 : SUCCESS in 25m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/58fe8003e0244e0790ffa0ad7767116d : SUCCESS in 46m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3eaf4e9cb3fe4331a1883a605c4532cd : SUCCESS in 1h 08m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4f06ef41f8004d248636e6a05c33cf39 : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1cc32b063f7f4485b7fad5152ca03ae4 : SUCCESS in 1h 29m 52s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/3f51148ed06543039663960fe6de227d : SUCCESS in 2h 03m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/764bc99650db4625adf565cd8a651db2 : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/95336ccc510442bda8dce690834534d7 : SUCCESS in 17m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8bdc2d33ca4411ab7094d23cae6e48b : SUCCESS in 59m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae4f13bebb21444c9bb859ed7eb4503a : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80e23afcdd9495da907de554edac230 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceca6e10d13d420da2c2e6cf134e3fb8 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ca8403ac044949a15ee8f126e2a402 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396f587ea3de47d3a37cb6ff31fb4ec9 : SUCCESS in 7m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/302c2b8b89634f85840d6419fa2322ac : SUCCESS in 21m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cd517ce903f441e95de1fd9c536bd50 : SUCCESS in 19m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/931d2dd1ad1846f8a1565679cf092480 : SUCCESS in 1h 09m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d859721cff7843f4ac4b9e1ed48a411f : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/975c2b645fbd4582a497eeda9fa96ca1 : SUCCESS in 1h 02m 38s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/02eec609c39d4005840df6ca064cb424 : FAILURE in 23m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/6661271e2c0c47a48989c136b1b1ef10 : SUCCESS in 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5583540cd46f2d7d4ef760b50e0e3854c2088a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 18:50:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e05b15a829dd6e306603b51d64ce2fc7b5de32a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:01:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92d26c6c629d329f3f4568b668dfa63fa10a5da3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:01:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/051e85a0ef474460b60cab9bc490b3a4 : SUCCESS in 25m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/66a4c52526cc40f5b2d5209b7ea5ea42 : SUCCESS in 1h 01m 00s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/85e7e754459b46e7a85f65d40a0950d1 : SUCCESS in 1h 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d0256ff7754eb99548e3788fea7e01 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d6d925bd9446668db4c466da6bdeb9 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df7c7b1eb004ba7ac4c7694a917cd21 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85347cb0efa443dab0b8244740a7f68c : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1449af8ced04ed0b5a62b58e7cd8e58 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1bd88eb945f4e9bbe35376a2c4fc26e : SUCCESS in 23m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/809557e209bc49ffb5d4f1a805acbce3 : SUCCESS in 51m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7333bfb35f648a1ac915699899fe5ec : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"363c14072a38e9135b772e8e6195e016f9fde3f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:03:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de68c4f80ad429689afd826900fc9ed : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"faaa38f9c4320e90a8ecd199f7610ed343e25f74","revisions":{"faaa38f9c4320e90a8ecd199f7610ed343e25f74":{"kind":"REWORK","_number":1,"created":"2020-09-30 16:46:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/755346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/755346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/755346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/755346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/755346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/755346/1"}}},"commit":{"parents":[{"commit":"d033d91f94a81b3566b489c46860af25fed8312c","subject":"Remove kolla-cli stuff from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d033d91f94a81b3566b489c46860af25fed8312c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-30 16:45:03.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-30 16:45:03.000000000","tz":120},"subject":"Fix Neutron migrations order","message":"Fix Neutron migrations order\n\nIt seems I mixed up the order when commenting/uncommenting.\nNeutron base has to go first.\n\nChange-Id: Ib2bf51d4c6e2a574b98671a699777cd8a9a22768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faaa38f9c4320e90a8ecd199f7610ed343e25f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faaa38f9c4320e90a8ecd199f7610ed343e25f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
