)]}'
{"id":"openstack%2Fkolla~755470","triplet_id":"openstack%2Fkolla~master~Idf2c504fe1f7035d7de77feb042480f6a1e16ee8","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-03 16:49:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf2c504fe1f7035d7de77feb042480f6a1e16ee8","subject":"Fix version-check.py idempotency","status":"ABANDONED","created":"2020-10-01 10:01:41.000000000","updated":"2022-05-03 16:49:11.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98d261bc9f5c2f509bd94eeff2332815f74de77a","_number":755470,"virtual_id_number":755470,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-01 11:07:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":-1,"date":"2020-10-01 10:03:16.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 11:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 15:56:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"}],"messages":[{"id":"4d90407614b47321ccd426196eb15d721e996caa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 10:01:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634367a28e6a34db7a29ce71c43489b884e17ddf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 10:03:16.000000000","message":"Patch Set 1: Workflow-1\n\nStill unsure about this","accounts_in_message":[],"_revision_number":1},{"id":"a17acc185a736cc2e6c21f71543009bb73801117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:07:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/84258f6ff3f549f8bbf55e9abb27d883 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d8d38e3d3aa406eaece2edd40440b43 : FAILURE in 18m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c0eb24f39ed46b8b8e424037893265a : FAILURE in 18m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/623c801a616a4f71bd37642023c3a72f : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/93be763e97844d6fb9f0d71d06f68096 : FAILURE in 22m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/621adce3f20a47569cf5df89fe7e275b : FAILURE in 15m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c93099297934b61bf9572726a92194a : SUCCESS in 23m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5920395abde6438fb162846562f9a8bc : FAILURE in 15m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e40e69cd34a148c0a3952117e0f67d30 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e152dfdc00ab45f1b0cc3bc48fb30acb : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61cfb6ddccac43dfbdab101313c2a2ec : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5170b370f2664cea89fc5591e661aef9 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54394534da0427fb1daa140f3aeb228 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1cec7cf709f4d46bc9e6d647776c9b6 : SUCCESS in 23m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7de74703825a46a0aa5b717555798774 : FAILURE in 17m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcd5914aab1d4d58a0322e8e136c4746 : FAILURE in 12m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42ae7aece9664f71a5104db2b40632f0 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0fdcc20fbb4c4636a5b228ffae2bb03c : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c1be3d20e3af294c52faa73b6f6f0dbca8d68309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/191e8bdc994e42c998559c9ec85e05e6 : FAILURE in 1h 02m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/248afed71c6741408cffc4f14a82d34e : FAILURE in 42m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9f36380ac9046a9bed320711a272594 : FAILURE in 28m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f991c228b6e54065adcd1562a18dcced : FAILURE in 1h 05m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e2420379e529150ff3fb3bc2facda8b7fa09b4f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 15:56:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98d261bc9f5c2f509bd94eeff2332815f74de77a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-03 16:49:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46f808cfe96d6c2e08d76322fc1356b39cbe2710","revisions":{"46f808cfe96d6c2e08d76322fc1356b39cbe2710":{"kind":"REWORK","_number":1,"created":"2020-10-01 10:01:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/70/755470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/755470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/755470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/755470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/755470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/755470/1"}}},"commit":{"parents":[{"commit":"d82afd5569de07e6571d79c62aa0e4923583d8e3","subject":"Switch to Victoria stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d82afd5569de07e6571d79c62aa0e4923583d8e3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 09:57:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 09:57:15.000000000","tz":60},"subject":"Fix version-check.py idempotency","message":"Fix version-check.py idempotency\n\nPreviously, version-check.py would keep adding stable- to the tag each\ntime it is run. This was caused by the regex greedily matching stable-\nas part of the project name rather than the tag.\n\nThis change fixes the issue by updating the regex.\n\nChange-Id: Idf2c504fe1f7035d7de77feb042480f6a1e16ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46f808cfe96d6c2e08d76322fc1356b39cbe2710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46f808cfe96d6c2e08d76322fc1356b39cbe2710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
