)]}'
{"id":"openstack%2Fkolla~757542","triplet_id":"openstack%2Fkolla~master~I26ad81a7076bad7a659e2d80a1ad0453c13794d1","project":"openstack/kolla","branch":"master","topic":"rally_openstack","hashtags":[],"change_id":"I26ad81a7076bad7a659e2d80a1ad0453c13794d1","subject":"rally: Add rally-openstack","status":"MERGED","created":"2020-10-12 11:44:57.000000000","updated":"2020-10-14 17:00:20.000000000","submitted":"2020-10-14 16:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757542-1602694729339-6db76d6a","meta_rev_id":"4e450d947cbfb0c5660478fde9585ea1da437df2","_number":757542,"virtual_id_number":757542,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-14 13:05:29.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 16:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-13 08:49:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-10-12 12:41:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-10-13 08:49:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-10-13 09:36:16.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 12:41:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-13 08:49:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-13 09:36:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-10-14 16:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"325006d2796b1d13a86f7fe06c6dea45bded0da5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-12 11:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39dcedc60a26f49150ab21f72e7f033978644d3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-12 12:11:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dee014180b46fef00b60512ea7dada6552f99351","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:15:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7a9ff75277646f6a7de3768c05edf4f : SUCCESS in 29m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e86551fa9e134dc19ccd32e9c494288b : FAILURE in 24m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6870eb11f8ba474ab41ac78f417621f3 : FAILURE in 20m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc3d5e42cd844a058f8eed8ddab5d00f : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4027c60ae6177d7ac9ca0d8bdc79820d38c5ede5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-12 12:27:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b97e75a142eaf85973a3a2db8843c63b644915a2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-12 12:41:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67942b0922f9676291863f42afd898edd9fe79aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/66042b1470934d869c354a57efc6d080 : SUCCESS in 26m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/76cc1f9d29d740b7801e3275410ad023 : SUCCESS in 26m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c23b504108f49bd8b5c5fa81019905b : SUCCESS in 1h 05m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/14860ce5ca454298b79e31202ebba0dd : SUCCESS in 1h 06m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f7ee9b84d21a4191a9b3acca732044bf : SUCCESS in 1h 11m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f697b0cf28a64e019239d25bbf2ab6f4 : SUCCESS in 18m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42b3bb59804a4832adde7b2ff4dd966c : SUCCESS in 17m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/72db997bee6146189e13d7bfd042d802 : SUCCESS in 1h 01m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4ea5f1ac8ed4fffae98f8cb88cb7ac1 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd9267d6d654a86953d4d47cf87fdb5 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6313d21d340e4f819e09ebece26adea9 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/092525611e534a4cb90d1cf01cd98e8e : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3499bd2b1f1a4954aa92f81e1b15cf7e : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04071860cecf4db4af0f5fedb181acc5 : SUCCESS in 21m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9b3daf6631240ec854c40cc006a58c1 : SUCCESS in 28m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/880cf9fb8964401ba7f587a93f780425 : SUCCESS in 55m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbe4928fa60b4a2b884c765ccab99a65 : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c89d453177a4da29ef529d940b49d34 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"da42d16d15ad77f9dee2a340001026f7c5b8ec64","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-13 08:49:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47c9d0dc5171b436bd5907e2330a44fd9fdf6a24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 08:50:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b955d3f8e9c24c5c53262e75d722365fa622abd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:14:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d82f895b314c4be2a2d34e279dc76913 : SUCCESS in 25m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4588761a3d8a46e593adefb14df3cce0 : FAILURE in 26m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8a8a8afc708948f791f0de164814f0fb : SUCCESS in 1h 17m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f40b7f4bf9f84161ad7e187a31c5a835 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e597e2e282431ba135b833666447b5 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/935ac7ccb71742da8b42abe33dc3bc7b : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/845b8a45dff6456a80083bfd060c1770 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be9e7dace744ee8b660f0236336bc5d : SUCCESS in 8m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b989de12bc944493a116931142bd47ff : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b1a1d40d81c4b33aaf3bda9e48a7fc3 : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/798be772985b4a8286776a8914a1a24c : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2c0818cf03b07e3e40252f2b98a1cb634b9c5c08","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-13 10:17:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95b4d8bbe2002dc6783c266f6ffa832d959b03f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-13 10:22:55.000000000","message":"Patch Set 1:\n\nuhh - curl: (7) Failed to connect to opendev.org port 443: Connection timed out","accounts_in_message":[],"_revision_number":1},{"id":"1af2729d673fd8cc250db1c82171b6af694e8ee7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-13 10:22:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b6d15c4a9111714c899e4f83fa04d2f71fe3ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:55:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/df2087d231f149259d30c6bea0a141a9 : SUCCESS in 37m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9de62a95ccd84c0f8e5c7085451e46db : FAILURE in 28m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebb62fe406ea496784ed5313ff8ffc7b : FAILURE in 20m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/30abdd2c8ebe40e4aad1ace68bd7248c : FAILURE in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ebc796e1cb03bbdcecd303c657cb0802319af94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 13:01:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fed202d829144086be089ffafe28d99b : SUCCESS in 21m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1660dfce61534f1fbf01c01bc84d8f92 : SUCCESS in 22m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/997a380a455c45218dba867841c3ae8c : SUCCESS in 1h 06m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9b4824da3cdd4ef686603735e66e3c7d : TIMED_OUT in 2h 03m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8e1b802d36374dc3b2e71d3443eb9f4b : TIMED_OUT in 2h 34m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5a1fa54cc1f40488dd5c0c6f1a35c0b : SUCCESS in 20m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a9ebbea5f00421ab725d69404253904 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/204437ce96a042fd92577bd35996bec4 : SUCCESS in 58m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7394a99e4b14f62b441082d028ed9f5 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6b943784df4a829e290d2f77ee733d : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35944d1d79d34e5f98761dbeed11ab8b : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f0ec5bd762143e790b278d970fa80c0 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a9a82fc3bf4979b51a224791974800 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e2a1de47d5944a49c0963467fe6895c : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2efa84d0d96848db9c7eb9634ef3a98d : SUCCESS in 19m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f2e5203a07f43b1a1d7b4d4021ae366 : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2c93a105f1e46d895ea7f64b7f3b344 : SUCCESS in 46m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e3da1ff29974d8c89b63d1e44197c25 : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cf74fa796dee2b2a4dcf2a49314ae4663d430ea2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-13 13:09:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a909f83c645888f4fe0d36dcbb1fecb967083970","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 13:50:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5ed1da2867c4010b0af248f18230d3f : SUCCESS in 33m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/622e71ba07724ea781ad4ff1b7930424 : FAILURE in 25m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4a0f80fa19247b795a6204abf436bcd : FAILURE in 20m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f8f944505f647c1a7cb2ba352d3e663 : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"badc519a948fc175515de1d3e33eadb273359191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:24:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/506a5fd937834335aaae9799ac368640 : SUCCESS in 27m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/996f24c75f2e4a34b0ebae9b4cc7d5c3 : SUCCESS in 22m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1a24686f9a2e41e0adf6d0efb96de07f : TIMED_OUT in 2h 04m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8be5d7c70a1248e88ab73c1bd8e6e85a : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0088dbf98c234f13b87f369bb2872bbe : POST_FAILURE in 1h 36m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7366e9bff1a04c028c6b55878d278c98 : SUCCESS in 42m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/156ac3ef0dec4e5c9d102e460dc1ef46 : SUCCESS in 17m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a93de8c58bb4a98b4b92ccb0bfe77a6 : SUCCESS in 58m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca255ab1188347499909699f30fc53ea : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025fd2c7fbdb4981a3d216e9d4c8afc9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23ba016765404c8a94bdb327d3c27dd7 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76d5132707b949548806e04d016c7565 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31470722ee74381968e6343d481a237 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ff2889746e44cd19e5db4449ce4676b : SUCCESS in 20m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/18778e0077064942ae925268a1dd79f6 : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a19c832d99f045289bf71d5a0b9e272b : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47552fadbec04b62b17e3eb229b42cc0 : SUCCESS in 47m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/486a7f94b01d4e799495ceb2a3c0a65c : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fccb02607043a805edb60c797089b652eaa71eda","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-14 08:09:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56a38db96c876e31557cbc59ae17ed304a03c40e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 08:43:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/93cdddee57ea46ec96064d5c1baf213b : SUCCESS in 32m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65599dfc1eed4afc815d989ee141276f : FAILURE in 26m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dc026039ffd436c958c6f03a24b2a09 : FAILURE in 21m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e1c851eff814e819758cb271b186839 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8237c1a67479f91c3816d686fc85a22d4ee5664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 10:48:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/928dffee84f0431da4667a77c26dd889 : SUCCESS in 18m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3503931c021b4edc8d36dd4aeb534d2f : SUCCESS in 27m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d0cb5b77c6fc4895b996e24c931d212f : SUCCESS in 1h 10m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b04f998646b9430480c89a4f40f357d6 : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f68e9c8de6964d5592fdfd19a1e2a256 : TIMED_OUT in 2h 33m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e215dcef55bd4363885f8c5d62399cee : SUCCESS in 18m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1d101b6809b45409525591cd68cc3ac : SUCCESS in 17m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/57602a9763a14209b8cb7bc7f7eb99fa : SUCCESS in 1h 20m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d9a1b536ff4ea28623798fabf9fb8b : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8319cd341a2c4e2fbed5f0f20b0e126e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12f5c43ae00f45a9bfd44337c4c35b0b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02ada4b6db5a451091a24417b1791397 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1415caa6fc7642ebb6bb5bc8084248b0 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5403315e3b144f24bc8c049bce880ad7 : SUCCESS in 18m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/080730977b1940f4bd136a470cfd25b1 : SUCCESS in 19m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/100a8c21aec14487878ed3947b3dd26d : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22e0d5b5585244a49024f0de4f296c56 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d5b9e5ed1464780a58af4867f4e8cbc : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"78e62e207b4bca046f8a70ab14da53a31c7d6f9f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-14 13:05:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1715772df206f82062ce6f344a8df9c6155abae6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:38:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9411a5f94e6947f2a193682bcdf018b0 : RETRY_LIMIT in 2m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84e948fdcad34e59ab02d679a8d95da4 : RETRY_LIMIT in 2m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c442a3b869048b892af19c99c816b90 : RETRY_LIMIT in 2m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bf7d4edbc2c40e8bbf2aebe36d9f925 : RETRY_LIMIT in 1m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07952a26196fc49354c6b601a02c7b7807c8a984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f34acaa40fe54ae5924175bfc3e1f3fc : SUCCESS in 44m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6a7a7e292c4b41279e8bc754a9739f65 : SUCCESS in 22m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f83ef1427d1e47c28d50e5df0c1c6c15 : SUCCESS in 1h 04m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3adb4d48da0f44478bc61a180f9398f2 : SUCCESS in 1h 18m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0a4b552aebd0466b8e0a8fa11f67d5f5 : SUCCESS in 1h 13m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ddf659343e04f29985d305a65b5eacc : SUCCESS in 2h 29m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e39d3883109442509ee01501d8feee4f : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb40e11743dc41d0b060d2cda7447a89 : SUCCESS in 59m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cc29ccc4f4048a7974a3cd58be43d22 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1c29f306f343b587c5d6f1e9429b29 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70aca35ac14f453bac8bc1b5a4739abf : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b99d1b2f038a4b37aede4e719521172b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de156e3d658648f0841835b56408dd84 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/922af6a5ffbe4075b22f6b208bbcbee4 : SUCCESS in 19m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa77e8f4a3fe4b5a8762ffb14a511162 : SUCCESS in 19m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca3c5950ad6a4fdeabf2aa2aaf3b8f50 : SUCCESS in 47m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb5c034fee2d4355ae9696c9100edeae : SUCCESS in 52m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8aef79934dd74d088c3a3f4872c248ef : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":1},{"id":"42e7df59a1712209e63e9cda14d205bda30185d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:40:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26a783f2232e432004ca9854b717202134e253a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:58:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/aa153eec380a4dda8ba5dc8ede17fa96 : SUCCESS in 25m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5c7fbab7af84474faf8af40ce7645476 : SUCCESS in 1h 07m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/29a87198028646bb882d09e9598afac0 : SUCCESS in 1h 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/773c455d68c647009c6ac768f925cc92 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296d871b871e4651a4918d4008e4ae28 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b46552555b448399aa3124c1afb84ac : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c208c307e4c4e859ea2482563cfad6e : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdd887609a6467cb88277df5cd9feff : SUCCESS in 9m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/460c234fb34d42a2af9e59c9536a777a : SUCCESS in 19m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46ef9844606c47f4ba0edf478764dc0a : SUCCESS in 54m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/16a05dd237224053bbacd3d3774de870 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2315d59871899a2109f3c4e122e2a94e5c36df9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1373a64cbb34ce08628fc53f977027c18b747533","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:00:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6951bbbc8aa425698cf110e5eea42cd : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc1abb3d7b217e46eb147bcc38f766c130bdd4cc","revisions":{"bc1abb3d7b217e46eb147bcc38f766c130bdd4cc":{"kind":"REWORK","_number":1,"created":"2020-10-12 11:44:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/757542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/757542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/757542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/757542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/757542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/757542/1"}}},"commit":{"parents":[{"commit":"524a821577d568a7a19983a7b6adc36e78fb9e4d","subject":"Merge \"Checks for heat_user_domain explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/524a821577d568a7a19983a7b6adc36e78fb9e4d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-12 11:43:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-12 11:44:54.000000000","tz":120},"subject":"rally: Add rally-openstack","message":"rally: Add rally-openstack\n\nCurrently source builds and centos-binary builds do not install\nrally-openstack. It results in errors with finding openstack plugin.\n\nChange-Id: I26ad81a7076bad7a659e2d80a1ad0453c13794d1\nCloses-Bug: #1803968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1abb3d7b217e46eb147bcc38f766c130bdd4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1abb3d7b217e46eb147bcc38f766c130bdd4cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
