)]}'
{"id":"openstack%2Fkolla~757566","triplet_id":"openstack%2Fkolla~stable%2Fstein~I030bbe32bd489c3031b065f74de83c658356d4b2","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I030bbe32bd489c3031b065f74de83c658356d4b2","subject":"Bump versions for Stein","status":"MERGED","created":"2020-10-12 13:33:42.000000000","updated":"2020-10-16 12:50:13.000000000","submitted":"2020-10-16 12:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757566-1602852447238-eaa21e4c","meta_rev_id":"6a3d07f38f9fc6d83f758426ede810e359c6cbbc","_number":757566,"virtual_id_number":757566,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 12:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-10-16 08:39:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-10-16 08:39:27.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-10-16 08:39:27.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 08:39:27.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-10-16 08:39:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-16 12:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27bf89fdc67e19327c54efabcae73dd95625a6ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-12 13:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60ee178ec4f4589ef02f74673c8f16496cc4c453","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-12 15:46:10.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027d say +/- of a fix:\n\nINFO:kolla.common.utils.bifrost-base:\u001b[91m\u0027ascii\u0027 codec can\u0027t decode byte 0xe3 in position 25: ordinal not in range(128)\u001b[0m","accounts_in_message":[],"_revision_number":1},{"id":"9f5b40ac8c4ab24cdee93570bdd73f5ea7c13207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 16:53:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/24868702d10c458485437ce77c63455d : SUCCESS in 42m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/55cb5633f58140f8ae86ebf2a41dc90a : FAILURE in 45m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/13a5df3d74cb4c189af8934c1f6da143 : SUCCESS in 1h 12m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2f4f81fdd6142699fe4611858328b24 : SUCCESS in 1h 05m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/74851b4e15e040cd92209ec5822ed0a8 : SUCCESS in 1h 28m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b219d716a078482193b49aeb870917f6 : SUCCESS in 52m 56s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bf66f524d6bc4597aab6b1ca715c435b : SUCCESS in 38m 28s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dc8311f24a734a89891d29b3aa88288c : SUCCESS in 56m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da0bd68b11314e2db22e193d7b10b47e : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7fe69c7f91849ab9081a45c8f0a1b11 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7423ac7f909e4034b769ff59d616f9e0 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c39d83bb7fd743718c673ae6ff1d77e1 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1e3c03b008746a4a69aae2446f9ebf4 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2254d10fed0344c2aa15aeb5dad45fc4 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8024b04de2794739addeac8462e7d540 : SUCCESS in 34m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40951839e90940f79bfd6e713d69c666 : TIMED_OUT in 3h 03m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee2c3222ccc141099b7976515d8ba718 : SUCCESS in 1h 01m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4dbf893856d94e269d803d449e7cfc8a : FAILURE in 30m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12762bbc4bf4405bbc718a29cc3ed8d2 : SUCCESS in 1h 01m 57s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"192c18a5af5292b24c2b5a7ff03138a16d0c0db4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-12 18:26:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9087279e822abeda541530d924208edea002712","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-12 18:26:19.000000000","message":"Patch Set 1:\n\nDoh: didn\u0027t release bifrost or update tag in kolla.","accounts_in_message":[],"_revision_number":1},{"id":"0aec387ca7b13137a301b7ceabd298f225962ca2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-12 18:26:30.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d54383fb8e654fca05216cf4b00e60559d9eda83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:50:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2bf6ba2eade042d0965964c906d1943f : SUCCESS in 37m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4d6c8ff1896f435492151655eaba1c12 : SUCCESS in 51m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/56a590712ce54af8b1ee51f07f4af300 : SUCCESS in 1h 07m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2d2656f687fa4afc87270a5430751014 : SUCCESS in 1h 17m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/27eed46d8d1b4b87b953712ebfb0d420 : SUCCESS in 1h 15m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/55af849c21a74aeabb7001234dbee477 : SUCCESS in 45m 08s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6ea2615ea0c548f8ae55ac621b4f13b5 : SUCCESS in 41m 48s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0bf69d12a79740f2a294d902b72d8fe3 : SUCCESS in 50m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44273ee59d8f447e870a23784ad01f2e : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fcbdb0d8db94208a4b4b8ec0f6fda01 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/776f860fe4534f2ea63920f6a0d4c350 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c91a3c36634e46ceb70f07696d558cc9 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c655ab81228409fad1a31f6242725b8 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab6a8b805e4496898412509be524c71 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a5e561160df4b34bde26b4ef1ab41f7 : SUCCESS in 22m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d7b9f75aec3497eb702668f6619d7c0 : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/beba4e89d381426ca26c39834a6561fe : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05a8e921e93b4ff0beb526e6d397b175 : FAILURE in 31m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af4c69ffa9f3468aa5eb79dc65c12a2e : SUCCESS in 57m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"aa0900260f3e843b6350488623d6a293c71e03f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:24:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"51ba3e34d8e8045ec7bf55ef459133565ed2fd5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:29:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1c141b3a265246f4a6e0fa85b58c20f7 : SUCCESS in 51m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1a751df126794ce89fff01965011ffed : SUCCESS in 49m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/42ed0335891942748c7b03855579bc58 : SUCCESS in 1h 23m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1f3f328237bc4fe8af6a17b027deab69 : SUCCESS in 1h 21m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a48a9bf410064afd9792562441d8c51a : SUCCESS in 1h 14m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d5faf6e99be14523bba268f899e41b89 : SUCCESS in 51m 58s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/32b20f574fbb47bb90469a53eed47584 : TIMED_OUT in 3h 01m 24s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7bc53819ad1048ea9958fb5bea9b8441 : SUCCESS in 42m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd394aab0652444abe61548a6c1bef0a : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98c4227cbbf4bdba015d24749a22ab3 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c02222702b84ea081701c03dad7e586 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58a3b3c5379e487ea8aa3dfc567310b7 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/852693cd970449dba99aef37d1f285dd : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c94845057e43cbbede0c21fcd85c56 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/539b3cc0a850450e9cb445017097855f : SUCCESS in 23m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/65854e51991441798a88217ecfcb1e77 : SUCCESS in 23m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3240d54da58540f48d16ea5c79e0468d : SUCCESS in 50m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3057cceb640c4ad6ab502aed52135153 : FAILURE in 33m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29bc3d67ffe843a5a5b9c8cb73bc4327 : SUCCESS in 1h 06m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"ec355856ed92b812ff97187dcc52494fb2390cc8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-10-16 08:39:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d178bf499f278791e1ab494d22057bf1d972d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:39:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8879e0d87b6beca7e5831b5e5a9d4518e105fbf3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-16 08:39:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7cd9bf9cbb59530ff42ab2b0ac3ded64907ad222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:47:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8ba6940fc6924fdcaf937ad6e9b2b03f : SUCCESS in 40m 35s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/623abb6d20574a8087b17f34a8347009 : SUCCESS in 47m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6681d6a578e44a54ba8f6c964c776d91 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67389307d5e463898d5f4d2e0ff3c56 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f789cfe9caf4aa883d702c947ab060c : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/029d6edb8845497d87df0645d70bf250 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6933eaef0b3e46cdaedf723198879b74 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5f30bd25e04fa59102f302859bc3f9 : SUCCESS in 6m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15f4473f22d64a52b03b07e438eeb778 : SUCCESS in 29m 44s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"8f4a00f62f54667f8409815349461861bcec0017","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:47:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6a3d07f38f9fc6d83f758426ede810e359c6cbbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:50:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b174ab996b204ce28e77da9a55533e87 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2036628594fc90e617dac0aa55339428bc62921","revisions":{"4bc297e1df1e0cf2db55dec086f0536d95d7e260":{"kind":"REWORK","_number":1,"created":"2020-10-12 13:33:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/757566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/757566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/757566/1"}}},"commit":{"parents":[{"commit":"c6835b12a6f1e8f715aa5c69383a8d1836551496","subject":"Fix kolla_set_configs --check with a directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6835b12a6f1e8f715aa5c69383a8d1836551496"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-12 13:32:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-12 13:32:20.000000000","tz":60},"subject":"Stein: Make Bifrost buildable again","message":"Stein: Make Bifrost buildable again\n\nThe longstanding Bifrost issues in Stein have now been fixed [1] for\nCentOS.\n\n[1] https://review.opendev.org/#/c/749014/\n\nChange-Id: I030bbe32bd489c3031b065f74de83c658356d4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bc297e1df1e0cf2db55dec086f0536d95d7e260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bc297e1df1e0cf2db55dec086f0536d95d7e260"}]},"branch":"refs/heads/stable/stein"},"82cb3229ec4b69c2c6edba9420d64cec93f9997d":{"kind":"REWORK","_number":2,"created":"2020-10-12 18:26:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/757566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/757566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/757566/2"}}},"commit":{"parents":[{"commit":"c6835b12a6f1e8f715aa5c69383a8d1836551496","subject":"Fix kolla_set_configs --check with a directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6835b12a6f1e8f715aa5c69383a8d1836551496"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-12 13:32:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-12 18:25:45.000000000","tz":60},"subject":"Stein: Make Bifrost buildable again","message":"Stein: Make Bifrost buildable again\n\nThe longstanding Bifrost issues in Stein have now been fixed [1] for\nCentOS.\n\n[1] https://review.opendev.org/#/c/749014/\n\nChange-Id: I030bbe32bd489c3031b065f74de83c658356d4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82cb3229ec4b69c2c6edba9420d64cec93f9997d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82cb3229ec4b69c2c6edba9420d64cec93f9997d"}]},"branch":"refs/heads/stable/stein"},"f2036628594fc90e617dac0aa55339428bc62921":{"kind":"REWORK","_number":3,"created":"2020-10-14 09:24:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/757566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/757566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/757566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/757566/3"}}},"commit":{"parents":[{"commit":"c6835b12a6f1e8f715aa5c69383a8d1836551496","subject":"Fix kolla_set_configs --check with a directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6835b12a6f1e8f715aa5c69383a8d1836551496"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-12 13:32:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 09:23:38.000000000","tz":60},"subject":"Bump versions for Stein","message":"Bump versions for Stein\n\nAlso, make Bifrost buildable again. The longstanding Bifrost issues in\nStein have now been fixed [1] for CentOS.\n\n[1] https://review.opendev.org/#/c/749014/\n\nChange-Id: I030bbe32bd489c3031b065f74de83c658356d4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2036628594fc90e617dac0aa55339428bc62921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2036628594fc90e617dac0aa55339428bc62921"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
