)]}'
{"id":"openstack%2Fkolla~758051","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I036cfe3af3dd96625f986a820c78a10e6558b086","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1884484-stable/ussuri","hashtags":[],"change_id":"I036cfe3af3dd96625f986a820c78a10e6558b086","subject":"Add sysfsutils to nova-compute ubuntu source","status":"MERGED","created":"2020-10-14 09:36:05.000000000","updated":"2020-10-16 13:13:18.000000000","submitted":"2020-10-16 13:12:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758051-1602853922306-31c078f6","meta_rev_id":"531638537c610c4948472aa755a76303a34c565c","_number":758051,"virtual_id_number":758051,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 13:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-10-16 08:17:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-10-16 11:23:59.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-10-16 11:23:59.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 08:17:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-16 11:23:59.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-10-16 13:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"caafd59641ad37a4f486f2d7eaf6be18de76aec9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:36:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0deb34cbfc7b8058c8d1f86d367b06711ca4e9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 11:44:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9f368c21696e47b8968bbcaab5bd363e : SUCCESS in 20m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4236b6083b0a44f4a913360c626c312d : SUCCESS in 27m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9d52d82320424cc79c57ed10d35b3b3a : FAILURE in 26m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/544ca64305be4411a0041911dc0bf5c4 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cdc2ca19226d48ca8b810f581004a074 : SUCCESS in 1h 12m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b1d1c8ea6f694b84a1e1783b2ee918c3 : FAILURE in 24m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bffe8d654b834c10afc269df50462f2a : SUCCESS in 31m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3acf9dbb04324d1aa3a378fbe679b25c : SUCCESS in 30m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ca75cd6eee5741869364344f11c57121 : SUCCESS in 1h 00m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c47f7394f19144d2aea568183b304fbc : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64d5587b22e40178259852f7fb4a83a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8bacc0b5a9d4541a181a93d640559fb : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed675ffd307745e2afd6dfc7f67010b7 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e37cea3fd241fa90f9516de011d001 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc311435217416481aa4e69197f577b : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03519012b3f24d7c856177d06be17c8d : SUCCESS in 25m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e8a97ce8ee1405d8b558750d9e1f80b : SUCCESS in 27m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5827db41f9e4816af025012097f5025 : SUCCESS in 55m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f85ec5a2232417ea08e36fe7478868c : TIMED_OUT in 2h 01m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bee2d57a45c4c0cb6ecb1654bb82a3f : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f64867aa158ecb9f6eb068778d727f2083bfefb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 11:48:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d8649c598e34211914a8213ccdadc1c : RETRY_LIMIT in 2m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c774a315570448a9eaa585c85dbe7a5 : FAILURE in 1h 30m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72dcfe65845a4f35aac276aa8aabbc4c : SUCCESS in 2h 06m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e34b17d42e194b34aab4d8127864ce57 : FAILURE in 1h 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a4fcec455a8c9025ab34b82387853fa7aeed60f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-16 08:17:43.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4315af70ed97ca301b1175d4836207b658e5f53a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:38:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4346c4b93be746019feb4b03ad47956a : SUCCESS in 32m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/859af6a65a1a45d8b043a73bf91c4ba5 : SUCCESS in 27m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d3753a378624c26821182698fb63849 : SUCCESS in 1h 20m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebaac72588f74c058d3a0fd15369cc64 : SUCCESS in 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04104ee34a0cfb471fa379339ee42ffab313be2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5cb9d9d459814a5d81481bd29aad756a : SUCCESS in 23m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/469d977060c04d809b955747b5450fe5 : SUCCESS in 1h 46m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/97038cc02d6744dd851e9b937cda49fa : SUCCESS in 1h 10m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b19c292ab58643dd9bc192e7251f16c9 : SUCCESS in 1h 04m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1d81ffb6ad2b495889e5602dffc0416e : SUCCESS in 1h 15m 12s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e91283efbce84d7e8c9b7d184b772fe5 : FAILURE in 22m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/538f4cbfe370423c9171fea2c9b9d5dd : SUCCESS in 20m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1434c1ad437e4c2faab3f3c3e195ff02 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e204d521a0c7403db21493cb1f427923 : SUCCESS in 54m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf6646490fc34d6e90daa316a4b45230 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5831ab9d2104b6da0354197b74347c4 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd068f2ecd924e3e970e67e59e6af08e : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc6d136fecb44275b45cac0d510f6021 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f738655701fc4d1ea974129821508477 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bfdf63d387c41b3bdd71c52683f976e : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb1a3a6e1269490faed5825147180ed3 : SUCCESS in 22m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/28e60d7f08eb4799b0a7bd402abc8e50 : SUCCESS in 23m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e7daff85c2f41d6b9f04b6d8e0dfcf4 : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93c10962561d446598fe0821a98f7b1e : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae3cd00cf5c44d89a6dc463aa7e0d93a : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bb5fc1e6d825512cb201d5228becbaaa21a4060d","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-10-16 11:23:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8aadd9a70ebcfc79538314873f76d129c40159ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 11:24:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3081ce4008ffe784c74920623993e27bb4b08ad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:12:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a0e13b831f04497fa3366ab7e18beb44 : SUCCESS in 24m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c9ed51bb04304a61824dfa48b8cf3c59 : SUCCESS in 1h 07m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/381a67f2e93e410c9ddbd563708dba1b : SUCCESS in 1h 13m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67fdd722eb8f495caaf64a8da64136f7 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d531988fbb3e43d49fe3296d349a1ed8 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8968705dd65c447198b362ba143c360b : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88d5ed7e870149bdb3f09478649ef306 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39aebe7ef0f74d348e4bfb8b0d59ee3b : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e080c0ad99244cfbd5d8ece3a9b241c : SUCCESS in 7m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64e89bef43fb4d91940218fd08ee4231 : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e69a1c71604a49d0b41a630af3a98cd9 : SUCCESS in 1h 03m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f36f4e4adc4a43878025299443cda837 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a428aa6c7f8ab25ed5f855f8d8406a11bd14f0d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:12:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"149eb47c8cf3319bdc54740fc30ba3e4167a5c7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:13:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1165b5ac913342c8a9467c1cf9dfeaaf : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0380d638f1c34f759f7437caef531582 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de7f3c25918439a97a9ab20b6135327938e74ca1","revisions":{"de7f3c25918439a97a9ab20b6135327938e74ca1":{"kind":"REWORK","_number":1,"created":"2020-10-14 09:36:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/758051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/758051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/758051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/758051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/758051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/758051/1"}}},"commit":{"parents":[{"commit":"c9316259647733ee1d35fd4613db916b4fedf46b","subject":"Release note for using stable sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9316259647733ee1d35fd4613db916b4fedf46b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-22 09:05:32.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 09:36:05.000000000","tz":0},"subject":"Add sysfsutils to nova-compute ubuntu source","message":"Add sysfsutils to nova-compute ubuntu source\n\nThis fixes the FC Cinder backend usage in Nova.\n\nChange-Id: I036cfe3af3dd96625f986a820c78a10e6558b086\nCloses-bug: #1884484\n(cherry picked from commit 0c8702c96f852eae7a1fb784a260c01c15fb86f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de7f3c25918439a97a9ab20b6135327938e74ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de7f3c25918439a97a9ab20b6135327938e74ca1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
