)]}'
{"id":"openstack%2Fkolla~758057","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Id0e738044a1931f9d611a7281a48ea4a593f1cf1","project":"openstack/kolla","branch":"stable/train","topic":"bug/1893204","hashtags":[],"change_id":"Id0e738044a1931f9d611a7281a48ea4a593f1cf1","subject":"Improve pip install process for offline deployment.","status":"MERGED","created":"2020-10-14 09:43:49.000000000","updated":"2021-01-26 11:56:39.000000000","submitted":"2021-01-26 11:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"758057-bug/1893204","meta_rev_id":"dbde333829a9cffc435951c12996a6610f413f94","_number":758057,"virtual_id_number":758057,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 11:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-24 19:30:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2021-01-25 09:59:51.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-01-25 09:59:51.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2021-01-26 11:56:39.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-26 11:56:39.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 13:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-16 12:00:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-24 11:34:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-01-24 11:43:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-24 11:49:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-01-24 11:49:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-01-24 11:49:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cac64b0c5927903a7d5cfb3d4b9cf020cca17f76","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82aada1e4c040c4f3ebe66a7bb4441255c7701c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:04:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c4bcbba6a1a342ebad87679594521d94 : SUCCESS in 35m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e7cdad1506744d92a82af706ba870003 : FAILURE in 23m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cd4701f655a14de7a3f59b69a5dec69c : SUCCESS in 30m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/41158104fb6449679dc3dff201185e5c : SUCCESS in 27m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c5ea253f543e47d0889f0b3346b384a1 : FAILURE in 22m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6a3f8721d9ea484c9b5fad7ca2c009e8 : SUCCESS in 1h 05m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fcdfabb7b1f447f68f61570e8682c997 : SUCCESS in 1h 13m 09s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a1527dcd978845e4adff8865435a68bb : SUCCESS in 59m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6424a2a0490d4e34a8242b006fe2b986 : FAILURE in 20m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7342376bb1674152a0fd8941a6470a92 : SUCCESS in 58m 05s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/dd1f2a50e4b746369750de0d2ac149af : FAILURE in 18m 16s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/0ef238ada9a94fcfa5226c8a13770f2e : SUCCESS in 33m 54s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/2b658329f52e479a9771fabf1ccd8c35 : FAILURE in 24m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/19bf75241e7f432c9190f73da5ff47a8 : FAILURE in 24m 34s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/8840c89bfaea4c97b9bb6568c1b3eb2f : FAILURE in 23m 26s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/609d66dbc3df453180923e4e6f12eb3d : SUCCESS in 1h 23m 56s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f5125500721b4143a596a7e9c7b05bbf : FAILURE in 1h 03m 52s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ac871e9227e242dcbe4fd88d676a0674 : FAILURE in 58m 25s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/17825b6fce244474a063172323634c54 : FAILURE in 20m 29s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b7e294b913664767a69e9705e9f0f019 : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3275cd8f0ffc485cb04a402848ebcff4 : SUCCESS in 20m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/67b13a848be044edaff038e1a5ee4f8b : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0d7c2baba2b46a1bda8a4f90edeb145 : SUCCESS in 1h 48m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21c0a2a71739474f875482077c3dfe48 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35efa3abbdc64691a7eb49423828981f : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bc051fb17c748c99b4e8de75298bbcd : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3f7f85ac8d64e00b5c3b385a7f796da : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bb6722ebb224805b348b162456bf374 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6950b9d3d34d41718faf2300485896a5 : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91e233ff8da44ee39c4b5900cd294e19 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9dcdb92ada184898aee6e7026f2f2568 : SUCCESS in 22m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ff9f2156b7b455fa3192300ab01cd0e : SUCCESS in 52m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebaa904b01f0444d9885e2240fe1861d : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e51204d29fb4573be2fda760dbe9282 : SUCCESS in 1h 04m 17s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/f9b0eb5f5a0d4bc3b58791fe48f70800 : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c5420285b6024dada76210865c535f03 : SUCCESS in 1h 29m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/14b86da133774b0d96d5b2841476fc16 : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3178bccbf76548b0b8c97e97a936cb87 : SUCCESS in 51m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"62bd88d63031faa3d310cc181fd73d4ef6cdb5ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-16 12:00:28.000000000","message":"Patch Set 1:\n\nwe can\u0027t really do this with c7 so might be best keeping ussuri+...","accounts_in_message":[],"_revision_number":1},{"id":"9ca1dbf57f05fc696e1563791790d91de5531c47","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-21 14:21:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"c36191e68143386e014a442435b2b529f6f0d8d5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:34:12.000000000","message":"Patch Set 1:\n\nWe need to get it working. get-pip.py is now Py3 so CentOS 7 is busted.","accounts_in_message":[],"_revision_number":1},{"id":"2420b8505fb29243059f453e2f7008b2a6ec61d3","tag":"autogenerated:gerrit:restore","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:43:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a2f9ba8814dd44764e6e667d2af778efeb139c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:43:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e47a3e55b72523b2f0357b7181ed6c61c3d7f8f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:44:50.000000000","message":"Patch Set 2:\n\nWe should upgrade pip before first use as 8.1.2 used by CentOS 7 is ancient.","accounts_in_message":[],"_revision_number":2},{"id":"fafd966a284ea3ed96cdc518748e21d42d150e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:47:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb1aef71354d03459b3706edce73153b756ac00f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 11:49:34.000000000","message":"Patch Set 3: Backport-Candidate-2 Review-Priority+2\n\nTrain only change as we are on Py3 elsewhere","accounts_in_message":[],"_revision_number":3},{"id":"d211c52edc1ec325e0a236725fc4781944350875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 12:22:03.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b4c20c7bc2815ab5905ccb666ac1fb7eb75d0b5e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 14:13:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b18bcaa39c4db13dbee1cec8ced9601e3a99c80a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 14:15:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a95c7607febfc7687fb7f74c025982af5f4b548d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 14:41:25.000000000","message":"Patch Set 4:\n\non py2 we can also use: https://bootstrap.pypa.io/2.7/get-pip.py","accounts_in_message":[],"_revision_number":4},{"id":"5f31ba4420d92064aae762aaa98fb73c17eb6724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 15:18:37.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9a84499ebc6a15de15f336ae75232836efca86e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 15:19:13.000000000","message":"Patch Set 5:\n\nlet\u0027s see how this simpler version fares","accounts_in_message":[],"_revision_number":5},{"id":"90fc21b7024c037dd2b97b2a828053de8300602b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 15:59:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1b14de2f59c42d65f7cb72a6c597eb5fa68f5f80","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-24 16:15:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"607e7fac1b02b716e42a2e942857116283e9ef4e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 16:16:05.000000000","message":"Patch Set 6:\n\nIt seems passing now.","accounts_in_message":[],"_revision_number":6},{"id":"270ed1dc0e8f16df38c86d87e7bc53d11d5d61e4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 16:19:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bcebc940e1a5ac9941e7a618b139496c69a5653d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 17:15:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6cffc0868a7f466eb5a3fcf67584da65 : SUCCESS in 37m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c80fa1ad459c436fbf8e4aab980d053c : SUCCESS in 41m 40s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/35979a114da84dd9aa92eb2023aa8a21 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fee3a3fa86204730a2c03566041841e1 : SUCCESS in 29m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/23cca8f122ff4e109450ccef392cbe70 : SUCCESS in 1h 05m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30678e0886094bcc9ae1686776010c70 : SUCCESS in 53m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6ffd5a264790404ca77b1943d364f65e : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/41de6178845b4e8580a0c7adcc0b53ba : SUCCESS in 52m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/864ce9c6d4164069aba187ab3876c2e7 : SUCCESS in 1h 14m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4ca4e6478a28467a9959b5a50cdddbb5 : SUCCESS in 46m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a09741b401b44ff9992cf6119508e168 : SUCCESS in 23m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/21026c2d3cae4ec9966e649a4eee8a6b : SUCCESS in 23m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6bb8e0034bb74651bd7e2e294f33b97d : SUCCESS in 48m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3321254f55d849b799d7620d33c813cc : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d147887eea974498a30d33fc8aec40c9 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43fc173298124c86a29fa5133a003708 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dec82033987449238a05ea732fdff3b7 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26af68b3c2042938058205e4058ab0e : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/038332343ede4c6196cb1b309464798b : SUCCESS in 20m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3bc2642a10ab4e2a81f8e7e5cd36a3d7 : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/43658418551c4fa5ba6469aa627555f2 : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a69429adb4114bbfb1284ba358705932 : SUCCESS in 46m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/78f6f64de18846a9af98e70967a525b4 : SUCCESS in 1h 07m 12s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"8883258f669b7f46593092c8bef237e8c305099a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 19:30:36.000000000","message":"Patch Set 6: Code-Review+2\n\nI\u0027m +2 with the current shape of this. Let\u0027s merge soon.","accounts_in_message":[],"_revision_number":6},{"id":"a1a1d23339adfc26720c680bb94efbd909882524","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-01-25 09:59:51.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d8afc328a14af138bec71a1f6e29f582f7c70ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:00:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f8a2a09f2e3493b8dd5175c408b4122de0223c87","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-25 11:09:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1a7d1d5b43f33261ef108c09fc9b93ad3f93f874","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 11:49:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c56a6e92d887b4444dcd0843c73dfc58b9eba161","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-25 12:04:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d6657a6f5a108286145ef36243316a2e7ecbabe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 12:12:01.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4de7dc457dea4d8d9365888732f8c78e : SUCCESS in 47m 25s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1ed906e20d4046b5bad4d0e373afd387 : SUCCESS in 33m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9c26d246cd5c4e3b92d7771f6fd1f544 : SUCCESS in 1h 05m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/992dcc03784e4d8ab54505e6410f1dff : SUCCESS in 51m 56s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/058853f6f2c244bfbb49f94459c0697e : SUCCESS in 1h 24m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/88cc33a587f04cfe9912186c00e7cd5e : TIMED_OUT in 2h 06m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6709b0b7574011adb9903f5fa0d9e2 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66cb292038ae4e2890f0411ff276e529 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/659d6eb550ad4939bef2425773bb4321 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7267f699177249d69ca23c8a0d257178 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0d4a520c0841aa97cfbed0d54ae501 : SUCCESS in 8m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be5237a87780407f9952cc73767a19f6 : SUCCESS in 23m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f62bf89e3bbe4320b400fb2fccc696bc : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e85c121b29e4240b48d023d17ed85da : SUCCESS in 59m 10s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"a181e15dc5cfe35e633c4e7e61f3e195a5ab938e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 14:03:35.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9e854a9d62732328ce990f4291992cca44c930e1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-26 08:49:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"061ad2d587a1edfa6312a73f180f8719694cb376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:17:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0616ac72f8f54389b7382d802e504b46 : SUCCESS in 43m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/31717540537f4762b65797c99d417175 : SUCCESS in 49m 14s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/15332b1eb4bc4fbfbaf57e89ec10cd74 : SUCCESS in 29m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/43e5509c2df64fbd9364b5c304b69943 : SUCCESS in 26m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0916924a16b744b394717e1cde3a7e89 : SUCCESS in 1h 11m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0b3cd159e8764dadbe92e9f4c1dd103c : SUCCESS in 53m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5827cbbd50b7454ba76d7e55b6336da6 : SUCCESS in 1h 10m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7c8e21976ee344bf88cbc537490c3970 : SUCCESS in 54m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0f443c47447b467e92f30897c4d08a63 : SUCCESS in 1h 24m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a9e877b610b34dd584997a2e4f0e1065 : SUCCESS in 44m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2afc1cf536944d689d4cd7ac51ea236b : SUCCESS in 28m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4c381359f8da46df93fcaa1ee21a446f : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/53051f6211244c408373c70ee8c45992 : SUCCESS in 56m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5821eae214e44227a976ae0f019acc53 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd2ed8b3b73545fdb9a17758fda80a6c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/199844b81e514601b6b85d0e15f236e0 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab4e37c75243407b85207fb1a3813e0d : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a870c43291834fde9223f296a839af5b : SUCCESS in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae85a8cbac8b470b88168c2d0e601a54 : SUCCESS in 28m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cd5015d00aa4f80a3390a86fc52583f : SUCCESS in 34m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c98f4afab5740ae853e6869e76510bd : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37aa2df8b12b4c4eb7c68d503b61af94 : SUCCESS in 42m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28abbfbd64d244f3b60c627792b3878f : SUCCESS in 52m 00s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"695ac7a6dfe3713ab13068cc0bd0c094d2c818ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:21:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9aa7dc5af74222488f9a710ede26c543b9387d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 11:54:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/30c405dc5a094e62b0b38ac0eab2ea9f : SUCCESS in 42m 48s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/708a33ddd86848efb55cbf203d954e51 : SUCCESS in 26m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/64ec40e6c8254e40b388103988583b25 : SUCCESS in 1h 06m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/95ebd819f5914ea69a2a5ee5b920a3f7 : SUCCESS in 1h 00m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dce842afe1314becbc939d9ae57ec4ca : SUCCESS in 1h 19m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e3df08a624a1428f8ac2adac3c280caa : SUCCESS in 50m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad10b9c8c9f48a4b79e911c67930ab3 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/daeb6293682149b2a3ebbbcd3ce62eae : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8af83615f38345daa40e677221933f1e : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36e7ebb0af474f06b130d120ccd10432 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f8673593344cab8381d8ac1fbe45cb : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3775d13de884bd7a7b649b9be6015ed : SUCCESS in 26m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bba40cbbe1f42cd867138de057d804f : SUCCESS in 39m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e45b864ccd84444b4cffba4e31008d3 : SUCCESS in 46m 18s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"ffc408998f8be630b94d7a5c26f56eaa7acb5596","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 11:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1b524f13a10cb47035b11082aea3b3bb7747ef49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 11:56:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f28f6fd707a4cafb3733b2162cdb2b3 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"278893ef1d9bc8f7d013685d0676ce2458fcaaaa","revisions":{"15e736585fb92ad5d8deb441be32bb309796913d":{"kind":"REWORK","_number":1,"created":"2020-10-14 09:43:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/758057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/1"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 09:43:33.000000000","tz":60},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\nCloses-bug: #1893204\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15e736585fb92ad5d8deb441be32bb309796913d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15e736585fb92ad5d8deb441be32bb309796913d"}]},"branch":"refs/heads/stable/train"},"ac208337c5de2549cfae7c7c61001706b2b0e8de":{"kind":"REWORK","_number":2,"created":"2021-01-24 11:43:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/57/758057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/2"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-24 11:43:15.000000000","tz":60},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n\nFixed source type builds.\n\nCoauthored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nCloses-bug: #1893204\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac208337c5de2549cfae7c7c61001706b2b0e8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac208337c5de2549cfae7c7c61001706b2b0e8de"}]},"branch":"refs/heads/stable/train"},"3994106b433c91cd95918e33598b2c3054965da5":{"kind":"REWORK","_number":3,"created":"2021-01-24 11:47:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/57/758057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/3"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-24 11:47:20.000000000","tz":60},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n\n- fixed source type builds.\n- upgrade pip to \u003c21 before using it (21.0 is Py3 only)\n\nCoauthored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nCloses-bug: #1893204\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3994106b433c91cd95918e33598b2c3054965da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3994106b433c91cd95918e33598b2c3054965da5"}]},"branch":"refs/heads/stable/train"},"c166e3472f2ed52bafa09e22f8a811f4076bd88f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-01-24 12:22:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/57/758057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/4"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-24 12:21:59.000000000","tz":0},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n\n- fixed source type builds.\n- upgrade pip to \u003c21 before using it (21.0 is Py3 only)\n\nCoauthored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/772198\n\nCloses-bug: #1893204\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c166e3472f2ed52bafa09e22f8a811f4076bd88f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c166e3472f2ed52bafa09e22f8a811f4076bd88f"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"34396cba2197dd4d9a22c50fac87c2f7ced8ca7d":{"kind":"REWORK","_number":5,"created":"2021-01-24 15:18:37.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/758057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/5"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-24 15:18:32.000000000","tz":0},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n\n- fixed source type builds.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/772198\n\nCloses-bug: #1893204\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34396cba2197dd4d9a22c50fac87c2f7ced8ca7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34396cba2197dd4d9a22c50fac87c2f7ced8ca7d"}]},"branch":"refs/heads/stable/train"},"278893ef1d9bc8f7d013685d0676ce2458fcaaaa":{"kind":"REWORK","_number":6,"created":"2021-01-24 15:59:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/758057/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/758057/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/758057/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/758057/6"}}},"commit":{"parents":[{"commit":"6beff0cfc92f3df0ff7210d8388506a1f961a5c3","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6beff0cfc92f3df0ff7210d8388506a1f961a5c3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2020-09-17 12:38:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-24 15:58:13.000000000","tz":60},"subject":"Improve pip install process for offline deployment.","message":"Improve pip install process for offline deployment.\n\n    * \"Use distribution-provided pip\"\n    * \"Use python\u0027s pip module invocation method\"\n    * \"Install pip earlier in order to avoid multiple installation\"\n    * \"Remove pip_version variable requirement and call\"\n\n(cherry picked from commit b93c40a3635fbce72e48dc29120e50484cba36ca)\n\n- fixed source type builds.\n- in openstack-base: install virtualenv using upper-constraints\n  (just like in kolla-toolbox)\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/772198\n\nCloses-bug: #1893204\nChange-Id: Id0e738044a1931f9d611a7281a48ea4a593f1cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/278893ef1d9bc8f7d013685d0676ce2458fcaaaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/278893ef1d9bc8f7d013685d0676ce2458fcaaaa"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
