)]}'
{"id":"openstack%2Fkolla~758540","triplet_id":"openstack%2Fkolla~master~I26dd3a02e5e0131f52eac1380562b610eea3a36f","project":"openstack/kolla","branch":"master","topic":"rally-openstack-source","hashtags":[],"change_id":"I26dd3a02e5e0131f52eac1380562b610eea3a36f","subject":"rally: add rally-openstack in source - follow up","status":"MERGED","created":"2020-10-16 09:17:12.000000000","updated":"2020-10-19 10:08:25.000000000","submitted":"2020-10-19 10:00:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758540-1603101630192-eaa21e4c","meta_rev_id":"83f3adaeaea231bc9d68e32d4f2c9e9f1ba97195","_number":758540,"virtual_id_number":758540,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 10:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-10-19 08:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-10-16 16:45:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-10-19 08:37:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-10-16 09:18:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 09:18:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-16 16:45:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-19 08:37:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-19 10:00:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"654032be64aa90bdcc9ed3ee40a2d8bdfb345e50","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-16 09:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913110be7b7c1b3c1fdfc607efd20e5fcc3ab513","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:57:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3eae20db38fe4c63b938c3f4bc87c9ca : FAILURE in 39m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/036b844e931546298ec94f772e23b9c7 : FAILURE in 27m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d618fd5eaf640758654116d1976979d : FAILURE in 24m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b21240996ed94695bbc361bdc7a243b1 : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"926939bc79164e01af513a6c3212eb4de8107960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 11:53:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5fc313d960db45348d9718d74bc2a1c5 : SUCCESS in 21m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fc878e900a8a445b80b5609f02af8d15 : SUCCESS in 27m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6f47d0102f0c49eb821dd0482769d0b0 : SUCCESS in 1h 02m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0ef2d3073647486784425aef145e63a3 : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6dcef410ca264514af116912f7456055 : SUCCESS in 1h 15m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f06da87a428d4fd986e59dc9de171c0b : SUCCESS in 15m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6638ff29883e45278d2874a91caa2221 : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6c081b43b9f44788a559d4020199f466 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ace648c6ecab40c4aa0b8da9d8ffb536 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b90c6c074a4472e93e4853ae3bea4c2 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6006266ac380432281cfde9717b59c8a : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d44f17e2265d444abc37c002909d5a01 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe67d402b2b4c7d9768fac7593725c4 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eacd35962e247f0a5a304b59d9adabc : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b020de5a8f91408799694b173f513df1 : SUCCESS in 2h 32m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7a8998821674d56a46509adfdf67595 : SUCCESS in 19m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0abcc930431449c3a39dbb7991f85d34 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bb325d46cf047f5baa7979848007ddb : SUCCESS in 42m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fca701e1cb784157bc57fe3d42e805ee : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8ad4d1ba0e494c096461118263f5a9645a178db1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-16 16:45:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcb1e500613bde53f4e101d0e4e4991a00206c21","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-19 08:37:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad4a498011793d7298740b62f933f1cbfb729610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 08:38:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbc2315f3b55d3491a04b39c87bdd7ed8b13fe17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:00:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/26e447b7737f49a3bf38a8d61d56b004 : SUCCESS in 31m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5fbcef46d08c43d384ff26b0ed027f0f : SUCCESS in 1h 14m 52s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/67333e79d3f64fc7ad9688000fe182d6 : SUCCESS in 1h 16m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7dd1a5dd9f54ff3ad73370d457b24f7 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff7383de3af4a258cf6e2e66f89a7aa : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/602a9cc3ab1943378fb07659ca0fada9 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7994728ef817425d96eb347cf2755f7d : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132e7cdec0824e3a9b1694a85540c174 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc0a2333762d4ff08363b84d8949569e : SUCCESS in 6m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9497dd5611b948869c5e6ce3fa188fd6 : SUCCESS in 27m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f71028183b64c83be2508233dfc22a5 : SUCCESS in 57m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d65c15d80a4146969b2c8c2153412408 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5ceb3b317ad5880e23d17f59816e0d1e3cae6504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:00:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2391a4d4a77376d8d64f5c5806664df2accd2288","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:08:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b837dca157489a9763b5fc6578919a : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/544cc6ff10054f9cb15b2e6ce2145390 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48b81b86939e54044d4fa289f2f9e4e620cf44bc","revisions":{"48b81b86939e54044d4fa289f2f9e4e620cf44bc":{"kind":"REWORK","_number":1,"created":"2020-10-16 09:17:12.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/758540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/758540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/758540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/758540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/758540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/758540/1"}}},"commit":{"parents":[{"commit":"c26086b1bea35784a10d9b9682b7301175d27603","subject":"Merge \"rally: Add rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c26086b1bea35784a10d9b9682b7301175d27603"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-10-16 09:16:14.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-10-16 09:16:14.000000000","tz":120},"subject":"rally: add rally-openstack in source - follow up","message":"rally: add rally-openstack in source - follow up\n\nFollow up to change [0]\nrally-openstack added as plugin to rally source image.\nChanged rally tarball to latest available version,\nas rally does not build tarbals from master branch anymore.\n\n[0]: https://review.opendev.org/#/c/757542/\nCloses-Bug: #1803968\n\nChange-Id: I26dd3a02e5e0131f52eac1380562b610eea3a36f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48b81b86939e54044d4fa289f2f9e4e620cf44bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48b81b86939e54044d4fa289f2f9e4e620cf44bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
